)]}'
{"id":"openstack%2Fskyline-console~975626","triplet_id":"openstack%2Fskyline-console~master~Iadc5eb972608cad8a735bb039c4d45482694bb5c","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-09 07:48:43.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-09 02:18:49.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-09 09:15:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iadc5eb972608cad8a735bb039c4d45482694bb5c","subject":"Correct various grammar mistakes","status":"MERGED","created":"2026-02-04 12:33:00.000000000","updated":"2026-02-09 12:24:01.000000000","submitted":"2026-02-09 12:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975626","meta_rev_id":"70526c423bfe0172de7435deb54bc63f217c2776","_number":975626,"virtual_id_number":975626,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 12:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 02:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 02:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 13:17:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-04 13:17:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-04 14:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8642e91e2d51103a38e5fcd7582001d442ea2be1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:33:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3340dae5ebc869cfe6e728f2ab8a917e52ddbddd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:35:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c2a49650fe6179de32da37cf32acc2a9587c58c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:40:42.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2730f47e5a4f68933990b36528912783b0de0a7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:41:32.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"bbea3a18d7220b7db72f4c650d30a1957cc5bdbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:52:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"20882c895958a9b6ee13a6c5a325718495ea3e8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 12:54:08.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"53b83be8e1936782a5773ed388c06026ad0b1925","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 13:00:02.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5391168013757e0fc4e28591f0136e6f59864b0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 13:01:43.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"ab0e91d7e05f2611de871d59e15da2daade6275e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 13:12:38.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"2bea78cf938698eb697899c9a1d605ff1362502c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 13:17:06.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"e643c388f7c17a8f0719f42a669e1cc5917f1770","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 13:17:13.000000000","message":"Patch Set 10:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"5a6f6a2cb3fdd55e16f1a2e101cd38ded0ae7934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 14:55:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06242c881c464cf9a413e2b0fdb3c0b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0254706d643748ab9d7b7c2f122bb159 : SUCCESS in 3m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6c5f97b45d3a4772a3c7e37f8fc3ac5e : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5b5a4024d42d44ac94b2aeae99756552 : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/427394ff4845470c9a1f3ea9cf8496da : SUCCESS in 3m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a959ff52bf4d44d5806f688161554c8b : SUCCESS in 1h 03m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/815efca4b314470c8bfbc2dcc37c2491 : SUCCESS in 1h 21m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8217be87745045279f555f2699fe5a55 : FAILURE in 58m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cd16f13cf2834a5fbc170274a15031aa : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":10},{"id":"b8b60c0bfa47b78e98f22d004946ed713e3dbc66","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 18:23:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e6b93b9a54b8548d5e8983177c42018fdb7c3a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:40:07.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5885859ea128443496754f678a3c71f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e0c865877743c3b26ac0eca659eb30 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5bc831466f57471b819adce46aeabdb4 : SUCCESS in 11m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3d443992d83542699aefe0f0403f3d9f : SUCCESS in 6m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1b8a4f86499b4160bbd6acb4d6128dfe : SUCCESS in 10m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/efd2718685e14f6b91c9ccd0fe6c8594 : SUCCESS in 1h 13m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/924bdcfcdaef47d6b3db05de5e51f01c : FAILURE in 18m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bc56a40e4d0a44f7887fab3f19a69c74 : SUCCESS in 56m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0349ebc2333a403394138a5058457fcb : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":10},{"id":"3961ecb18f7ac095639e776f99d3f2ed979c44f4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 19:45:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f0bc028449ebc4c3fb7895ea59d6157afd01ac42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 20:54:38.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fae087d3a5c46f5aa755cd5cf6eaef0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2bc7f8d8d744baa53cc1a65bc7f651 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aa926cecd6a445f2bfe7d7b4b7899cf6 : SUCCESS in 8m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bf1ebf82f9f54edd8482b39ee6b47b5a : SUCCESS in 4m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8035c0807891478fbe4df521b86120a6 : SUCCESS in 3m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b3e213518779424eb06b6dcc7a9c6592 : FAILURE in 52m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e1ef1a46722548419e2f248d9aed244e : FAILURE in 16m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d55e425391cd454680cb284fc0549cc5 : FAILURE in 1h 05m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/eff33d9e5c77463a810a4703c8420694 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":10},{"id":"aa815c40e5cbeb239350cbd7bc1123f2187d823b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 20:57:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"71e8ccaf40e8d462a5aebfa9d56db1e1e80904cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 22:23:17.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b2c644d82ab43aa9618e79fb90862f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bac231fc844734b026aaeba8bf3cfe : SUCCESS in 5m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c797adab61f14581a5d8ea66d3680996 : SUCCESS in 10m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1ea02e0e449f4e4e96904b87331fbede : SUCCESS in 4m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6d48b4f0393c40bc8e443f29260896d0 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4eb095c46450496480e96b350fb98a50 : SUCCESS in 1h 20m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/255c54d2b7324d31a91d9b4c0e19e498 : FAILURE in 36m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bf579e8f6c884f6f8cc8c219157fa471 : SUCCESS in 52m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8c1fbfbad71741339c98d8dcab85e7ba : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":10},{"id":"5e000cfdd66a25736b9d4d21b75b0f655911a55a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-05 13:44:15.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fceeb43a0a9bc1353a3e54a65f106feccdf04c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:25:03.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cf82ca82d7e45c0a6d543f1a2db707f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9f5edc896142499ca8f18674605327 : SUCCESS in 7m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/487e4adaf75f444f8dcf380adcd3dbd0 : SUCCESS in 8m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d7a799035dee4edc992106e9d93761c8 : SUCCESS in 9m 05s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a4cec17127d045a1bc0a5d48ff065f04 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/84147b3f7b9d4754829eb617e09455c0 : SUCCESS in 1h 28m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/06c330cdb28b43bb95e7caea8b273c53 : FAILURE in 32m 07s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/54d360d2dfb74847adc523b2156ba405 : FAILURE in 41m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6327a60571f240e6b81b254993bc517a : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":10},{"id":"21198701fd66926510a15fae5db8a6153a6bc090","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 09:45:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"625da1dee16a081fcd18d17deddc3252ba3fe392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 10:58:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0a99772fb7b4878993666418b6e12a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5170f5e267534f89a40f8fa305e6828d : SUCCESS in 6m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2b921169ec0d45508d25ca229199e856 : SUCCESS in 10m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/14de076cc78f4d329938c16c10e5404b : SUCCESS in 6m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5f1121fac4f543dfbf3d9ac7617441b6 : SUCCESS in 4m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0329d57b70644e3886690a22e639f5cb : SUCCESS in 1h 02m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2e891df3b7aa4513a676ed3fab964c3a : SUCCESS in 1h 06m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f693c7f021894dbe90cebff9f4d2580e : SUCCESS in 49m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/62d822ccb98a4131a43d5b8ceab3ba59 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":10},{"id":"978f50cbaeced635cbd5f523ec4a898a3a1a6c4a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 02:18:49.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"ae8b509455d1e6215034c82bb73d602cac7f61b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:19:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b493b3d8957a8b955d8ea4c8fef82b9083517260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 03:59:55.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63e2e36441f54b9c8d19cf791ab46572\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a9d184e6f740a1938fd556f761c92c : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cc9d508e77964b4e911238c70e8ba1f1 : SUCCESS in 9m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6e03b6f0debe46958ec86de52f9f82c1 : SUCCESS in 4m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7597d6819e9f4de9a422155792bee896 : SUCCESS in 6m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/66fed0dd85a44eb3925e5fc2573976cf : SUCCESS in 1h 20m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fc10347b30f140be821cd7fb78d7c294 : SUCCESS in 45m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d06a4774b5744b49b4f499a9b81a9aa1 : FAILURE in 44m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/40aac4fc4f254effaadbd275f0e86a5f : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":10},{"id":"0a8abdded6bfd6f256e501859ccd9859f43fb816","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-09 07:48:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"450f118158ed788680d30366b6e51436d5000a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:06:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8d2ad1856124265a5c4e1459f83c6b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7222f70280784b00abba08a57c81b271 : SUCCESS in 6m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ca6934596a004d3d9ef48a8de3ce7f3e : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f78be82dbb25464fa6519af48d6410f0 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1f3f76db43c44276b10ef6bc0b295c94 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6a10a63101194a729f6f254af06c238d : SUCCESS in 1h 00m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/464c27260fe34fe19cbf22e0054ed2d2 : SUCCESS in 59m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d8847ff40eb74a338af7dab59d179503 : FAILURE in 40m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/55241aac0b9d4fcbb4e7ef89b461e654 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":10},{"id":"8d9fc47cf251b5767f3a2f3e57178c5e617e0680","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-09 09:15:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c3fa9a69d9046c829b70b291788a661ec9255ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:45:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72ac87d856e64a9f9336ff0533541590\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7008a1bf7dce42f8bcf530985d0d047a : SUCCESS in 6m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ef5962e6cc864b919d5c20cf4ec6c546 : SUCCESS in 4m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c8708ca0046d49149573e7eb4582ed6e : SUCCESS in 8m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/597f8b134a9146cf89ae6618e8dc0a65 : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1769709e4e82402780fcc5c0b57028b0 : SUCCESS in 1h 22m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dccb644ff9ad4a4f9ed8dc01065eb35b : SUCCESS in 50m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ef05253a7a5e40c69d4e4d06486cf619 : SUCCESS in 54m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3412c5691ac24d21a9ed93d86658e739 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":10},{"id":"ed48df8dfa9ba6cef5c975dc3c8bf5ed3b9c8afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:46:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e4f21f2964400e87f355b603115c82a3034bc5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 12:22:27.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d7c52257f34e3e8dbd5fb22b6db39f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c7bb0fbd78436987c9c4a1653cd00c : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/afaa190df76c4b659d15fc8d0ef7adfa : SUCCESS in 8m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/af53f17b51e5444884f3e44c06ff1f2a : SUCCESS in 8m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a1c60dbec7b94294a0ed349b194d54f4 : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/459e2aa497f1404cb4a35c9e74885443 : SUCCESS in 1h 23m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f9be48f40b21462986a08bd0670d58fd : SUCCESS in 59m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3d96ad1491604fae873025153e2ce970 : SUCCESS in 53m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c7a2bf7a493f4d42b5ef293b76bdedd5 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":10},{"id":"168e1362e7df43a4e00f320bc35c1ebb4d22be19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 12:22:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"70526c423bfe0172de7435deb54bc63f217c2776","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 12:24:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd41ed6c57c4a109ce708a12926429e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60c942c68684940b2805f84743579e0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7bafd87d6498e7a5c549add8f2ca910d752dd89b","revisions":{"c80dd96192d8aee1bb524788ac918d1cbf15d98b":{"kind":"REWORK","_number":1,"created":"2026-02-04 12:33:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/1"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c80dd96192d8aee1bb524788ac918d1cbf15d98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c80dd96192d8aee1bb524788ac918d1cbf15d98b"}]},"branch":"refs/heads/master"},"071c0d0cdb6a7cf6501b6e2fcde9b3bb01288188":{"kind":"REWORK","_number":2,"created":"2026-02-04 12:35:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/2"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:35:13.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/071c0d0cdb6a7cf6501b6e2fcde9b3bb01288188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/071c0d0cdb6a7cf6501b6e2fcde9b3bb01288188"}]},"branch":"refs/heads/master"},"4c7d3a3db2f15577b16af7b9446e17d5e19ff1bd":{"kind":"REWORK","_number":3,"created":"2026-02-04 12:40:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/3"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:40:41.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4c7d3a3db2f15577b16af7b9446e17d5e19ff1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4c7d3a3db2f15577b16af7b9446e17d5e19ff1bd"}]},"branch":"refs/heads/master"},"e437247fd491d13ff97582f98162e2f5775149f4":{"kind":"REWORK","_number":4,"created":"2026-02-04 12:41:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/4"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:41:31.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e437247fd491d13ff97582f98162e2f5775149f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e437247fd491d13ff97582f98162e2f5775149f4"}]},"branch":"refs/heads/master"},"2542ed5aa5724955c1ba5e3bc82216f1b02cd6af":{"kind":"REWORK","_number":5,"created":"2026-02-04 12:52:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/5"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:51:27.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2542ed5aa5724955c1ba5e3bc82216f1b02cd6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2542ed5aa5724955c1ba5e3bc82216f1b02cd6af"}]},"branch":"refs/heads/master"},"a99014455eb0337e50900374b812933af7b641c5":{"kind":"REWORK","_number":6,"created":"2026-02-04 12:54:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/6"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:54:08.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a99014455eb0337e50900374b812933af7b641c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a99014455eb0337e50900374b812933af7b641c5"}]},"branch":"refs/heads/master"},"281655b5b2dd4e610d2ac3cbda7d0076987a2d14":{"kind":"REWORK","_number":7,"created":"2026-02-04 13:00:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/7"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 13:00:02.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/281655b5b2dd4e610d2ac3cbda7d0076987a2d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/281655b5b2dd4e610d2ac3cbda7d0076987a2d14"}]},"branch":"refs/heads/master"},"58f820dc2d037ef0ddbb58840b7095f88c5e50b0":{"kind":"REWORK","_number":8,"created":"2026-02-04 13:01:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/8"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 13:01:43.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58f820dc2d037ef0ddbb58840b7095f88c5e50b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58f820dc2d037ef0ddbb58840b7095f88c5e50b0"}]},"branch":"refs/heads/master"},"7a6b54cc2ee314136f461eea939a36ac5e0927f6":{"kind":"REWORK","_number":9,"created":"2026-02-04 13:12:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/9"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 13:12:38.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a6b54cc2ee314136f461eea939a36ac5e0927f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a6b54cc2ee314136f461eea939a36ac5e0927f6"}]},"branch":"refs/heads/master"},"7bafd87d6498e7a5c549add8f2ca910d752dd89b":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-02-04 13:17:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/975626/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/975626/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/975626/10"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 12:33:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-04 13:17:06.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: Iadc5eb972608cad8a735bb039c4d45482694bb5c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7bafd87d6498e7a5c549add8f2ca910d752dd89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7bafd87d6498e7a5c549add8f2ca910d752dd89b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
