)]}'
{"id":"openstack%2Fskyline-console~975876","triplet_id":"openstack%2Fskyline-console~master~I7a18d3ffdef93937dae631a4e90a1a2498f4955b","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-09 02:59:53.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-09 02:14:06.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I7a18d3ffdef93937dae631a4e90a1a2498f4955b","subject":"Improve affinity description","status":"MERGED","created":"2026-02-05 23:03:32.000000000","updated":"2026-02-09 03:01:05.000000000","submitted":"2026-02-09 02:59:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975876","meta_rev_id":"f0873f54ee205fedae3e0e89af92124a0677b2e2","_number":975876,"virtual_id_number":975876,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 02:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 02:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-06 09:46:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 02:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 00:20:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 09:46:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-06 09:46:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"4c78a98d8e24793ebd63ab122b427b0fa3281b6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-05 23:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc21832730579eba6da805e17e3fa979435e7006","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-05 23:06:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"92d294eff349e4d856f63e916cc5f9c929a2d4f3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-05 23:06:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"dfe4018aadb06e372cd0b08354d969a803c9ce2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-05 23:09:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5e3b88bfaf83779e19f9658164e03b3f776e17dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 00:20:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/087153b5602d4f6f920f3cb305997329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59efad0241e4d86b36572478cfaf7e4 : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/42eb13c0736449e3b569cda341bbccff : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/99afc1ec1d16404281c0a163e887356d : SUCCESS in 8m 30s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d43e3889b91e49d8b71f3c116d60f2a9 : SUCCESS in 10m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/58d0b09506f34e74a40079a97a799886 : FAILURE in 1h 11m 17s","accounts_in_message":[],"_revision_number":3},{"id":"916679a9c4c265ef210ebf0d997739fd8ef2c8d9","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-06 07:57:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c354055305b55fe2ba2c4336b750828fe05431b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 08:37:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9b81c8ad9a462d9405b696b57a515a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0ee9667bc54e53b0cf41a760ad4a25 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/eee6128f774344688e8d7cb5c3a2e921 : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a8156cde5f014bb09c1c2068593eacf9 : SUCCESS in 7m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/426faa29417640009537fd0bc2359fea : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d2199963cb254d4980d14bdba6475865 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f074870dd323e3aca221a4b2b36ae508e4722df2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 09:46:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ed956d3c08e1a5097c42ef51a67534dba0c44a7f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 02:14:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ba3b5c5250c6d38126d698800483cc840a9c7080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:14:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb89e98cae97dded93e3f73dc06f13383bc5bcbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:59:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1699619869814ec5b751d2b24696debf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5059fd8f116144f499c1ab6f1c7c5c58 : SUCCESS in 4m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d8d058901f0444139b8525c60caed9d4 : SUCCESS in 5m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/62fcc80bacd6463b9e7f61d19f7605f4 : SUCCESS in 4m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2898dfe196984c768d2f0e6a01e832a1 : SUCCESS in 4m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/03ecd05d894b4a40ba3d6bd784b04757 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":3},{"id":"be7f15c8dd9a854ade5df2a64964068239064e28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:59:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0873f54ee205fedae3e0e89af92124a0677b2e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 03:01:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0917a0a3f044463cbecfca81b5d9ffa2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d40d42b5a134c6190f42e4b4760ef4d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2beacc3945eecd33cda0ab7e9bab0397702a09c8","revisions":{"786eef5710ff3c07344562ba5715e033f045aaa0":{"kind":"REWORK","_number":1,"created":"2026-02-05 23:03:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/76/975876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/975876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/1"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:03:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:03:32.000000000","tz":0},"subject":"Improve affinity description","message":"Improve affinity description\n\nUse more relevant label for choosing\naffinity rules\n\nChange-Id: I7a18d3ffdef93937dae631a4e90a1a2498f4955b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786eef5710ff3c07344562ba5715e033f045aaa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786eef5710ff3c07344562ba5715e033f045aaa0"}]},"branch":"refs/heads/master"},"294554692a0a8160837c978952faefc0e0766a78":{"kind":"REWORK","_number":2,"created":"2026-02-05 23:06:16.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/76/975876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/975876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/2"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:03:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:05:58.000000000","tz":0},"subject":"Improve affinity description","message":"Improve affinity description\n\nUse more relevant label for choosing\naffinity rules\n\nChange-Id: I7a18d3ffdef93937dae631a4e90a1a2498f4955b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/294554692a0a8160837c978952faefc0e0766a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/294554692a0a8160837c978952faefc0e0766a78"}]},"branch":"refs/heads/master"},"2beacc3945eecd33cda0ab7e9bab0397702a09c8":{"kind":"REWORK","_number":3,"created":"2026-02-05 23:09:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/76/975876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/975876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/975876/3"}}},"commit":{"parents":[{"commit":"398abbda519d33aa47d839caed5840d98965e151","subject":"Merge \"Provider Network Type - add possible translation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/398abbda519d33aa47d839caed5840d98965e151"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:03:32.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-05 23:08:57.000000000","tz":0},"subject":"Improve affinity description","message":"Improve affinity description\n\nUse more relevant label for choosing\naffinity rules\n\nChange-Id: I7a18d3ffdef93937dae631a4e90a1a2498f4955b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2beacc3945eecd33cda0ab7e9bab0397702a09c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2beacc3945eecd33cda0ab7e9bab0397702a09c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
