)]}'
{"id":"openstack%2Fskyline-console~975903","triplet_id":"openstack%2Fskyline-console~master~Ic31019062c07008a90d4ecdf1869f35e980ba992","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-09 05:08:42.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-09 02:27:32.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2026-02-09 05:08:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic31019062c07008a90d4ecdf1869f35e980ba992","subject":"Fix: Add fallback to prevent Copy Public URL error","status":"MERGED","created":"2026-02-06 12:13:46.000000000","updated":"2026-02-09 05:09:31.000000000","submitted":"2026-02-09 05:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975903","meta_rev_id":"f7ac8ddba3c35e26c2f7234d6849d166f966e64e","_number":975903,"virtual_id_number":975903,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 05:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 02:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-06 13:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 02:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 13:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 13:55:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-06 13:55:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-06 14:38:00.000000000","updated_by":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"e12b305a73760d4edc8a8bf42b1e94087a1a849c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-02-06 12:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8eb347b214bc4868612a38ad0b340f60f703aa60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 13:07:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c43f1c704ad4579a44da99397ac52f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b05c1050812471288808661de71c001 : SUCCESS in 5m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b395d66d3584ea1a9d30b57a1b0e64e : SUCCESS in 4m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4d98fa58a29b49bea60929e1405b62fa : SUCCESS in 6m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/712373458dcb4a148a10ff7985da7d0e : SUCCESS in 6m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1c8d498f25cc4904b31d015855f6db35 : SUCCESS in 48m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6d28555a5bf89762d69e0a146cf0e6a72922aa2f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 13:55:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6497636e0e550f0f5b34647b5d8d500fd62a17da","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 02:27:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"851d4f55a36e0c8356d4e5f2f360f14c0f3431a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93dd196a81610dff5b726c319121104437397d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 05:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cad56b378f040f794e87ef7957f2b45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc33c50070247fd87a63293f0cadc64 : SUCCESS in 3m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e8146996c38d4025893167fe2e0d0bd2 : SUCCESS in 7m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/57bd8876d4ce4c2580d4612fed1bb015 : SUCCESS in 4m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5aab61bdc3cb4da6a2ce7a61c80dd6e0 : SUCCESS in 3m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/208aea2792d74e69aa018c2225962e6b : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c9b2257d8f83f0369031fd553e57a994395391f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 05:08:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f7ac8ddba3c35e26c2f7234d6849d166f966e64e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 05:09:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1304802bb2854000a88ef9eeff7dd818\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e9946935e04a05aad48ad588e6c51a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c391c572b57a7bff19ddcf2c07fc23553b791dd","revisions":{"7c391c572b57a7bff19ddcf2c07fc23553b791dd":{"kind":"REWORK","_number":1,"created":"2026-02-06 12:13:46.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/03/975903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/975903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/975903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/975903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/975903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/975903/1"}}},"commit":{"parents":[{"commit":"bd5604c46b6d334e958603bb66d35f7e9f5a9b3e","subject":"Merge \"Object Storage - rename form, fix for \"New Name\" field\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bd5604c46b6d334e958603bb66d35f7e9f5a9b3e"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-06 12:11:57.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-02-06 12:12:16.000000000","tz":330},"subject":"Fix: Add fallback to prevent Copy Public URL error","message":"Fix: Add fallback to prevent Copy Public URL error\n\nCloses-Bug: #2139689\nChange-Id: Ic31019062c07008a90d4ecdf1869f35e980ba992\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c391c572b57a7bff19ddcf2c07fc23553b791dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c391c572b57a7bff19ddcf2c07fc23553b791dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
