)]}'
{"id":"openstack%2Fskyline-console~975951","triplet_id":"openstack%2Fskyline-console~master~I4125d3a9dc73490ef0325d6c434a666ee84993f2","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-09 10:00:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-09 02:40:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-09 08:39:01.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I4125d3a9dc73490ef0325d6c434a666ee84993f2","subject":"LB, correct various grammar mistakes","status":"MERGED","created":"2026-02-06 19:04:12.000000000","updated":"2026-02-09 10:49:41.000000000","submitted":"2026-02-09 10:48:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975951","meta_rev_id":"56d3c323cae7cc000fb6dc7a8818e250f48c51cc","_number":975951,"virtual_id_number":975951,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 10:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 02:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 02:40:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 19:32:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-06 19:32:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-06 21:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"365ccf43d3ebd4818d7444e1de6543185ae8558c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ebb4076403d378237985bef006ec93603e115ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:25:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"18e74a62a6f7e85a2c6878baa0f88320fe61e966","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:32:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2d2ee3e01961b1f2c28774d7c26cdd5d4c91cb35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:32:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6084fbc52f2438e699e26aed3a7212187f2bd7e4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:32:55.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"b7c83b884ccef7e67c37d6957764578a10bfcaa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 19:34:38.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"43e2a87794261f767e96010ae5ac69d811396bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 20:19:54.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"70376d7acc2cc659e0555a9d8ab4bc08136d5a9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 21:19:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec294f5508a4b4db3916bc392a477a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343e36f3e67b4d02bce987d427f33c35 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e3685f8a8de74f7da8440a4261ed13a7 : SUCCESS in 8m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0da91cd721034440a6e8ccdba25b4d3b : SUCCESS in 3m 44s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/da8d2597ed034303af22db3d57f5eff8 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e43186187a644e918e076880a3bd49d0 : SUCCESS in 59m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a028f515ba414054857bd94a1d677dc9 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":6},{"id":"af2b7f9c6de40a4cfa0609a74e47f0062a5fc0dd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 02:40:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7b628ce9e5457aff8fd13c1249e3c2f6cb47dc8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 02:41:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ccb0f569e35570f66f89ab00c30f2a7e2573621f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 06:06:33.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64d38b2a55204fc3a6163899043be65a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69112f8097d34b6dbccf35dd6f4fb0f9 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d9843b96d61241faa53a7d54e0665d81 : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ee5f1cc7d9344b59ae8ccb735047dde2 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8ec7846a01b741bc8f5e48eac33933f4 : SUCCESS in 7m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/75eb3b80627941e083e24947b4171170 : SUCCESS in 57m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8fb0447d5ddf430884442b9f2441f7e3 : FAILURE in 1h 05m 13s","accounts_in_message":[],"_revision_number":6},{"id":"5a86cac4d9757fbea7d1dfe479e03b1d76576ce4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-09 08:39:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd9a53881ec803199d9fd2941f7a595ff48cc8cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:42:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb82108b04f45a89f94ac94defe344a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b0170a0d194e3c8d82429f2beef2c1 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4737ad2f3d8640bc867de2864a1bb055 : SUCCESS in 11m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f103efae922a44e898fe136939af12bb : SUCCESS in 8m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9e8fcec74a4d46c085e5a85ef13a0780 : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4012daed8d7a401eb393c0df1b790fe0 : SUCCESS in 57m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/89a576f6504b482f890e83966bd5d786 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":6},{"id":"53d0e002064762c3420d352b276241cd01bfeae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:42:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6343f8ae926375343da652cd077c9b664fa489fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:48:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4861a3bd7a847c2a569b93ecc212a57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963f08b8d09d4cbab690502881d4f5cb : SUCCESS in 5m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/882153c591d2491aa47ec1b832c4c9f0 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/233cbfa45dbc410b860a6ffa31b02ea8 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3f488ed9be624273ba7d9db2bbebdf27 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4eaf362d363f448dbdc2be1b1cb1e9c7 : SUCCESS in 1h 01m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/36961e798c384fbfa30c046ed791e8c3 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":6},{"id":"85f107ff8e9e8045a4fc2e02e8bf962b48184bfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:48:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"56d3c323cae7cc000fb6dc7a8818e250f48c51cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:49:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/166cb0b4125f487d9f5951996bf499ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3806fefc790e45998b1ae32d3b55f715 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1c1076325db61ff3a95e8ce1af9b868240efc120","revisions":{"846b51b90eafacdb283ba4c0df7bef6ee75c1058":{"kind":"REWORK","_number":1,"created":"2026-02-06 19:04:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/1"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/846b51b90eafacdb283ba4c0df7bef6ee75c1058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/846b51b90eafacdb283ba4c0df7bef6ee75c1058"}]},"branch":"refs/heads/master"},"bbda92f6624e213a39d91d7f7a7bd1cea6a320c9":{"kind":"REWORK","_number":2,"created":"2026-02-06 19:25:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/2"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:25:15.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bbda92f6624e213a39d91d7f7a7bd1cea6a320c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bbda92f6624e213a39d91d7f7a7bd1cea6a320c9"}]},"branch":"refs/heads/master"},"920a258c3ec592a118642e685fa0a7aae1f9cddf":{"kind":"REWORK","_number":3,"created":"2026-02-06 19:32:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/3"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:32:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/920a258c3ec592a118642e685fa0a7aae1f9cddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/920a258c3ec592a118642e685fa0a7aae1f9cddf"}]},"branch":"refs/heads/master"},"e457f9db0399bb8418d0bacfedeff647fa4a583d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-06 19:32:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/4"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:32:45.000000000","tz":0},"subject":"Correct various grammar mistakes and optimization","message":"Correct various grammar mistakes and optimization\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e457f9db0399bb8418d0bacfedeff647fa4a583d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e457f9db0399bb8418d0bacfedeff647fa4a583d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"265a161600b7a5f37bf484760fc185766a628c65":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-02-06 19:34:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/5"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:34:38.000000000","tz":0},"subject":"LB, correct various grammar mistakes","message":"LB, correct various grammar mistakes\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/265a161600b7a5f37bf484760fc185766a628c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/265a161600b7a5f37bf484760fc185766a628c65"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1c1076325db61ff3a95e8ce1af9b868240efc120":{"kind":"REWORK","_number":6,"created":"2026-02-06 20:19:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/975951/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/975951/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/975951/6"}}},"commit":{"parents":[{"commit":"282239332959381d149fe3a68f3669f2a3ab6818","subject":"Merge \"Make Load Balancer table columns hideable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/282239332959381d149fe3a68f3669f2a3ab6818"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 19:04:12.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-06 20:19:54.000000000","tz":0},"subject":"LB, correct various grammar mistakes","message":"LB, correct various grammar mistakes\n\nChange-Id: I4125d3a9dc73490ef0325d6c434a666ee84993f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c1076325db61ff3a95e8ce1af9b868240efc120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c1076325db61ff3a95e8ce1af9b868240efc120"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
