)]}'
{"id":"openstack%2Fskyline-console~976074","triplet_id":"openstack%2Fskyline-console~master~Ic32d0a665dce9c5b6e724d5cb552c15033c0bd17","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"Ic32d0a665dce9c5b6e724d5cb552c15033c0bd17","subject":"feat: Update caniuse-lite version","status":"MERGED","created":"2026-02-09 03:26:06.000000000","updated":"2026-02-09 08:49:21.000000000","submitted":"2026-02-09 08:48:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976074","meta_rev_id":"c6866738d8585df08a60c446151c11fef30d6f3b","_number":976074,"virtual_id_number":976074,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 08:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 07:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 07:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 05:00:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d9465d57367d1bb8765ab5bef9c453c9cefbba8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 03:26:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e69df01a8eb7d3b34c01d8b1dce36e604846983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 05:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc813ebda3e41258303a734e147f3ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f33e3dd3db47768d0fb0339cf0a0b5 : SUCCESS in 5m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8d2f732a5a344b088b542e9c510ce490 : SUCCESS in 8m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fa4c35cd9fe84744898e8bd1fe481e34 : SUCCESS in 4m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b188212da6b94bc0ba64809d1830cdd6 : SUCCESS in 6m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fdff279511274043a676d82b5811ce3f : SUCCESS in 1h 31m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d91f0def3de2459f94dae111e04b8b3d : SUCCESS in 58m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/60e084d84e834ae6b30b993c7b3c25d8 : SUCCESS in 32m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/184098c5b5774d98904279ce078fb4ff : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"022173cfd43b30ae940e0c622aeba58ff5122412","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 07:38:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40d7d096660876c2acc1853625273ea219b34180","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 07:38:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05b4b6cd87d9ea2134ca5f40d77fde93d5d51725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:48:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af9741f3f9f747a98268a1529295da3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dddaa253bf794e3f8e039d926949001a : SUCCESS in 5m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e68f5ca297db4ce2b5304be6a57dd5da : SUCCESS in 4m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a0c83491dfdc42fe969fb2fb9578346b : SUCCESS in 8m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cb31ac60582c48708b3f67bc3a8e99dd : SUCCESS in 4m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5dc2daf43fb14082826ba9415dbdae43 : SUCCESS in 1h 08m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f3c5e64c46da4398814ad049d3ebcc10 : SUCCESS in 1h 01m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ffda33527c5d47fa9c92c8125d9c2d72 : SUCCESS in 49m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/39ecc67dc7ea4e3d8bf428be3813c77b : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":1},{"id":"65d135bbd9d50d3e43a49e19bbbe545be5830d4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:48:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6866738d8585df08a60c446151c11fef30d6f3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:49:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bd1c88752e84c6faf35282dba074a55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e29e13359d411187c9277158ef2b92 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9adcc80cf7ea71f63c946f80d479600123c6a6f6","revisions":{"9adcc80cf7ea71f63c946f80d479600123c6a6f6":{"kind":"REWORK","_number":1,"created":"2026-02-09 03:26:06.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/74/976074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/976074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/976074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/976074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/976074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/976074/1"}}},"commit":{"parents":[{"commit":"954d63d97335cc59c7312108992903805cb99428","subject":"feat: Remove the detailed logs during the packaging process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/954d63d97335cc59c7312108992903805cb99428"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:23:23.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:25:24.000000000","tz":480},"subject":"feat: Update caniuse-lite version","message":"feat: Update caniuse-lite version\n\nCore purpose of upgrading caniuse-lite: Sync latest browser compatibility data for accurate front-end tool decisions, avoiding adaptation deviations or redundant code.\n\nChange-Id: Ic32d0a665dce9c5b6e724d5cb552c15033c0bd17\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9adcc80cf7ea71f63c946f80d479600123c6a6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9adcc80cf7ea71f63c946f80d479600123c6a6f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
