)]}'
{"id":"openstack%2Fskyline-console~976078","triplet_id":"openstack%2Fskyline-console~master~I3ce86f4107559906a82b6acf0ee622f8da9c467f","project":"openstack/skyline-console","branch":"master","topic":"i18n","hashtags":[],"change_id":"I3ce86f4107559906a82b6acf0ee622f8da9c467f","subject":"feat: Update i18n","status":"MERGED","created":"2026-02-09 03:54:55.000000000","updated":"2026-02-09 08:50:09.000000000","submitted":"2026-02-09 08:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976078-i18n","meta_rev_id":"8094b27106214caeadc6731febc41bc4fd3ca044","_number":976078,"virtual_id_number":976078,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 08:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-09 07:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-09 07:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 04:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0b9e81911f5767ecfed69b84206eea8d3ea84c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 03:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55c4d09711dc7b1693cf81b432846f58f5a8b159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 03:57:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1932eb9820d8b7f11255e546604c9e1086cca61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 04:42:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8540bbfb93944f3ebcd4f8102bbd922f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a70a83c3cf4e09ba23b808de776fb8 : SUCCESS in 6m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bd72918888674c0bbb80cd2a1e09080c : SUCCESS in 11m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0aee38d39b124019bdac7b2e668a38e1 : SUCCESS in 7m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/117d3a70d82144dfb9935382db72f807 : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c8b12de6f1054edab64e111135582278 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":2},{"id":"87926a762a7a94b0f24943665861584b51877152","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-09 07:38:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a1498d4d5b113d9425b7b500330daa7af4bab0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 07:39:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9cdb789b2299a3e694f63dc73193f02f73108169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:48:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5ed3e9c2a7417c83a2112b8a8fc30c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed99a771b8c043f9ae5f08ca77567e72 : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/40a9f7c768bb465d88b81f775fdfe235 : SUCCESS in 10m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/304e4c043dee4cd1b99257381487c5e4 : SUCCESS in 7m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4ca50cc5a5104e4f9484db4b3b0de086 : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/514242a4eb4a4a5893a4cdc50173b916 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":2},{"id":"38880b944d097e16776815aea24d33935715d51b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:48:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8094b27106214caeadc6731febc41bc4fd3ca044","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 08:50:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ed02f89bc6402fa92caac300db1636\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b75060676c40c4899185c5eaeab948 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d36cbd7b8db459a98abdb61baf68e60611539243","revisions":{"850a9c29bebfb54b057f9a6545a20745e58c44d1":{"kind":"REWORK","_number":1,"created":"2026-02-09 03:54:55.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/78/976078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/78/976078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/1"}}},"commit":{"parents":[{"commit":"bca14dc814b6a8052ea411a830c62c3115e13dcb","subject":"Merge \"Improve affinity description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bca14dc814b6a8052ea411a830c62c3115e13dcb"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:51:40.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:53:01.000000000","tz":480},"subject":"feat: Update i18n","message":"feat: Update i18n\n\nUpdate i18n files for each language\n\nChange-Id: I3ce86f4107559906a82b6acf0ee622f8da9c467f\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/850a9c29bebfb54b057f9a6545a20745e58c44d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/850a9c29bebfb54b057f9a6545a20745e58c44d1"}]},"branch":"refs/heads/master"},"d36cbd7b8db459a98abdb61baf68e60611539243":{"kind":"REWORK","_number":2,"created":"2026-02-09 03:57:47.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/78/976078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/78/976078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/78/976078/2"}}},"commit":{"parents":[{"commit":"bca14dc814b6a8052ea411a830c62c3115e13dcb","subject":"Merge \"Improve affinity description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bca14dc814b6a8052ea411a830c62c3115e13dcb"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:51:40.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-02-09 03:57:27.000000000","tz":480},"subject":"feat: Update i18n","message":"feat: Update i18n\n\nUpdate i18n files for each language\n\nChange-Id: I3ce86f4107559906a82b6acf0ee622f8da9c467f\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d36cbd7b8db459a98abdb61baf68e60611539243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d36cbd7b8db459a98abdb61baf68e60611539243"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
