)]}'
{"id":"openstack%2Fskyline-console~976287","triplet_id":"openstack%2Fskyline-console~master~Iebdddb84a0a5785149f25fa9e03228c00c7ad67a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-10 20:51:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-11 04:17:49.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-11 05:45:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebdddb84a0a5785149f25fa9e03228c00c7ad67a","subject":"Networks, \"Shared\" field rename to \"Shared Network\"","status":"MERGED","created":"2026-02-10 18:50:30.000000000","updated":"2026-02-11 05:46:37.000000000","submitted":"2026-02-11 05:45:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976287","meta_rev_id":"af9ce8cb093f9564fb6abfcfba4635eb662cbf43","_number":976287,"virtual_id_number":976287,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-11 05:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-11 04:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-11 04:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 19:05:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-10 19:05:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-10 20:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5056bcb6df5b2bf81a14b68a89658b98da205cef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 18:50:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd7eb0ae16bdc8bbaea6597e730139e1650f001f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 19:01:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1b5058b71ebfaba5293c30a2b52384838e2352c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 19:01:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4030718add8fff6754c16818c2ed2c6a0726f324","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 19:03:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"53f1da82968a5e85acd2c865c9b4d77afa25ece0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 19:04:20.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6bce6ed0c067ddafdb79707ed1b0650e65c64809","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 19:05:28.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"5446ba1b57be83e4ec15f64d2fd37148e91bd18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 20:36:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a45a2fd63d5041588ee34c2ef2d922ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7730c57c5b74094a52858e229bca0cc : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/01bc62d1bdf548bab5c3182cc1c14611 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a13f28caaca44f5b8eee59da9b2c1fe3 : SUCCESS in 7m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ed6fa2d709c04c2b81e4fb9288598621 : SUCCESS in 8m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ec7e23684d634097817ff50c91a317c9 : SUCCESS in 1h 30m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/086b1d52c3634afbb0542755e8e9daf4 : SUCCESS in 1h 02m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c5f7620a300541e9a58ccc6c23928428 : FAILURE in 1h 11m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ab780ded374b4ea5b0b976b43715d420 : FAILURE in 19m 52s","accounts_in_message":[],"_revision_number":5},{"id":"113edfc72a92d8313f8334ffe3b4086499568e21","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-10 20:51:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc6645a3b539b7a78f51ad2e1d6c4d7862efb72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 21:52:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/907887ca65ab494293ea1b5ded16625a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733e4ce199c043d2bcf3ae658e648a1d : SUCCESS in 5m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a40a86c80efc4873bf1dc56865424aa2 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0fcdee0995344d088137a8b7f0e90101 : SUCCESS in 4m 46s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cf1ed0570e5440d69a7b6d48c9747422 : SUCCESS in 6m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/63aba9b416084caf940c417b06268ba0 : SUCCESS in 50m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dd2f5d52df364ee49c8379663289f9f2 : SUCCESS in 59m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1603569234074bfb911803a3f0510d71 : SUCCESS in 52m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/89d76c8e517948a980e78412b3e1ca9c : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":5},{"id":"0ecb3f1a428e4a0aa8448a017f8dea8e9e2dae8f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-11 04:17:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"774b48916252888bb46b0d42b26e0b5310f4e836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 04:18:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0feb346d4054b5996c80919757eb18757a952557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 05:45:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509f23311b574b23a23137f71c585750\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d309edcf3543fb81f3ce98bdced2ce : SUCCESS in 3m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c02cbfdd2a4a4a48b0cf58a152fe7147 : SUCCESS in 8m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cbd9d9e5ebc947f980e884acd39341fd : SUCCESS in 3m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/107767eef73d4c3d824b5826d64163bb : SUCCESS in 3m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/48221cef439d430abc7d0be38d8987e6 : SUCCESS in 1h 22m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a8c751423db54393acb4353091a8db3a : SUCCESS in 30m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/303dac790a8947a9935c40c6c00fa554 : SUCCESS in 31m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/23de971ca4d74c7d86f7b8a65e0b1caa : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":5},{"id":"dc79224d30b6e77a51bb9ba1f4a10b935d63a51a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 05:45:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"af9ce8cb093f9564fb6abfcfba4635eb662cbf43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 05:46:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1029293556a341cdb47bbe68f6d235ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3d5fded8334cccb501e085e15ca175 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1e1f56723d9f7df4d155b479562bdb45d557803a","revisions":{"120ad75c45b68ff428c40940d11570f6c7d7cce8":{"kind":"REWORK","_number":1,"created":"2026-02-10 18:50:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/87/976287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/976287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/1"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iebdddb84a0a5785149f25fa9e03228c00c7ad67a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/120ad75c45b68ff428c40940d11570f6c7d7cce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/120ad75c45b68ff428c40940d11570f6c7d7cce8"}]},"branch":"refs/heads/master"},"ade737b3efa4d33659a774b42513fba5d6a7ad2e":{"kind":"REWORK","_number":2,"created":"2026-02-10 19:01:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/87/976287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/976287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/2"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 19:01:08.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iebdddb84a0a5785149f25fa9e03228c00c7ad67a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ade737b3efa4d33659a774b42513fba5d6a7ad2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ade737b3efa4d33659a774b42513fba5d6a7ad2e"}]},"branch":"refs/heads/master"},"7dcc18f364d4d9b95b1ae7aabcfa6624f178e6ec":{"kind":"REWORK","_number":3,"created":"2026-02-10 19:01:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/87/976287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/976287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/3"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 19:01:53.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iebdddb84a0a5785149f25fa9e03228c00c7ad67a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7dcc18f364d4d9b95b1ae7aabcfa6624f178e6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7dcc18f364d4d9b95b1ae7aabcfa6624f178e6ec"}]},"branch":"refs/heads/master"},"eafa5b824c5c664055e63cb2bbfb4943d217f826":{"kind":"REWORK","_number":4,"created":"2026-02-10 19:03:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/87/976287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/976287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/4"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 19:03:09.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iebdddb84a0a5785149f25fa9e03228c00c7ad67a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eafa5b824c5c664055e63cb2bbfb4943d217f826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eafa5b824c5c664055e63cb2bbfb4943d217f826"}]},"branch":"refs/heads/master"},"1e1f56723d9f7df4d155b479562bdb45d557803a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-02-10 19:04:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/87/976287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/976287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/976287/5"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 18:50:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 19:04:20.000000000","tz":0},"subject":"Networks, \"Shared\" field rename to \"Shared Network\"","message":"Networks, \"Shared\" field rename to \"Shared Network\"\n\nChange-Id: Iebdddb84a0a5785149f25fa9e03228c00c7ad67a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1e1f56723d9f7df4d155b479562bdb45d557803a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1e1f56723d9f7df4d155b479562bdb45d557803a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
