)]}'
{"id":"openstack%2Fskyline-console~976315","triplet_id":"openstack%2Fskyline-console~master~If202a4792ed4272257101d30137befada9c26479","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-13 03:56:16.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-13 00:46:42.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-12 14:14:04.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If202a4792ed4272257101d30137befada9c26479","subject":"Hide Flavor column on Instances page","status":"MERGED","created":"2026-02-10 22:12:55.000000000","updated":"2026-02-13 03:57:09.000000000","submitted":"2026-02-13 03:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976315","meta_rev_id":"c81ca83cc20483341bdb706241fce1d138c504e7","_number":976315,"virtual_id_number":976315,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 03:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-13 00:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-12 14:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-13 00:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 22:37:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-02-10 22:37:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-10 22:42:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-10 23:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbd39f485c4e7c94eae6bb0fc154d62a25a0759c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-10 22:12:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b077cfbd343cf220890fa0099988ca1df708b8f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-10 22:15:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"10a64cdf731a24334a3d52ad13d7772a53341b1d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-10 22:15:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a343d4ab3170dee20c87e15e9b5f37e9b5e811c0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 22:37:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5528910d90abae18413114e018e429e449df7601","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 22:38:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa12679aa173c888d721cc7b6fc6422faff7f152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 22:41:34.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f6b5fa08793b25a27ded23b8294125aa10e31ba0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 22:41:54.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3f26cfb488a566071f02e25e8e5e7592966d94e7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 22:42:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8b6c8c2a54545c6f6afb3a2129e4b9f256c29a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:36:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecc284d93329460f81b8a878060b8c59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfaedbf9f6ae46a8949ef6ebb4c7cb9c : SUCCESS in 7m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/86b0a9d69c6c47848438222be21bd2ed : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/da169f92f0d848baad53935516436e14 : SUCCESS in 4m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bff36b3610e448608089686509b670e5 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a31bc3329ab1454488c4711189d1f43a : FAILURE in 53m 41s","accounts_in_message":[],"_revision_number":3},{"id":"40633cf19e83a827a51496b0098f0b355629e90a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-11 00:21:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a20aeb8b238750578120b93347012fdf17bcff6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 01:49:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f88c21ce7bac49d689a356002c69d61e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ffe4ee9eda4eb1ab5312b1c2f15659 : SUCCESS in 6m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/548be76e3af04c55a4be1cbcbaded01b : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6b410defc77c452781869643f6dd6511 : SUCCESS in 3m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/82942f31b15b4b8cab3e2ac486400de5 : SUCCESS in 9m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fe4bc688bcdd4ca7ae7423c22a08558d : FAILURE in 1h 15m 18s","accounts_in_message":[],"_revision_number":3},{"id":"65b8f8bfc71dd584ce6fd86eeef07a76854cac56","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-11 07:21:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb4fe51437b03289b9a7252e854f5cf2079eb1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 08:17:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bc2f2c25b3042718f3d57d1ccd51bcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe405377a064a7ab045d5439abc808a : SUCCESS in 6m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2792fb0d88514b7f97d63354113c6030 : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c5781fd5859b41a1b5cf8cbd354f5bf1 : SUCCESS in 4m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/556485f7ef754880ba6955f35c02cb28 : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6e56e1a52c8e40528308088859767b3c : FAILURE in 52m 32s","accounts_in_message":[],"_revision_number":3},{"id":"cf2e0e79b03a2b6884a8829a6da85e27c1ee6fca","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-11 08:19:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69a26beb2326c12fc9b14dd34a18ac75e445ff3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 09:13:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb911c5af19b41cc80921eab1f34bfce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57a06849f4744f2ba81f30f4a2c4073 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0f7c61eb01674b9c870c60c246a9adb8 : SUCCESS in 11m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3d15478961f3497d8e2797f769db08f1 : SUCCESS in 7m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8c6d9ce5b82e45138845f95c26fc4fb5 : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/60d06d737c8e4c71a37a5c5676bc1161 : FAILURE in 53m 06s","accounts_in_message":[],"_revision_number":3},{"id":"bdffae58bc8b648f46c245483ac7ae0a4274fe5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-11 09:38:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":4},{"id":"371688fb6ad64660f5f97753d5d63a2a04a7b7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 10:34:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94040a34a29a47c58aa5f41e23d42873\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b21c5d97be465180f54cfaa2c9f026 : SUCCESS in 5m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9e1e8ee062d8487c909bfba6f4232c20 : SUCCESS in 5m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/046092a2bb1942518c256acf627ce4a1 : SUCCESS in 6m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/aafe1517b9b54ce9a5781b92f7544e76 : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/03cf8b6e49bc4ee088124017299b11c2 : FAILURE in 53m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5c5e3aadb1ae60d8f3477505888073fe724f7f69","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-11 10:59:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"940bec189ff1f763f9a0c3330f41551cbb84311a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:14:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dae76e20df03496da49afa164e007a86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63848fcff5ff4a85b693c069a91d7400 : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/68f5c283db5f456da0dd4b9eb7dc77d1 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a258895230ef471b8f9f592086c54326 : SUCCESS in 4m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4383e18b32a94980b8bcc780a602efdc : SUCCESS in 3m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e12520a95c54461889e295f96962af8c : FAILURE in 1h 12m 09s","accounts_in_message":[],"_revision_number":4},{"id":"863f2de993a0b31ec167b1723d4e1ce11533affb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-11 12:53:51.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8ada88d7be60b16010926166407a3b72566e17c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:23:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c3f45b880a5441fa19dfed8225aca54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac1640548a34887adb713fc9ccdd4be : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5a12a59f74b6407a97c15ed0f686870a : SUCCESS in 5m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/27c97259d7fc4dbcae44c14b8300de73 : SUCCESS in 8m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2f3dfc0d405e47928c8ad99c11b7597b : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2a6eee24a65e459e9e06b837d2811cb3 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":5},{"id":"4b372c74a00191f855f9938bcc7ff9ad324f3bbd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-12 14:05:01.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8144e642c64a3bce4267791f141729a8dab6bd85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 14:07:54.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f68bdc675e19542f5bfa84894346a47295da8e8b","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 14:07:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e692227e9262ce016bae99c065efe9a7e0edf9c8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 14:14:04.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f7e4a87e38b6707fa65ec27f7aa712222b285865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:06:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13397b6c70f54474877bf676f20535c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00cfce7ffba42cb8a65ecf4b844d030 : SUCCESS in 6m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/493337f831cf415eaa25ba210d5aca19 : SUCCESS in 5m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5e6952ce81524969819c2dd13d32e97d : SUCCESS in 5m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d2fd95cd6aaf4c4cbf5aa3eeccf24063 : SUCCESS in 4m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/07ef70346435451a8799019ccb2a19ec : FAILURE in 55m 20s","accounts_in_message":[],"_revision_number":6},{"id":"c37616af6f094dd807175aee4b61f74c660d5984","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 16:45:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"07369e7bd4b3d4f70094186f257bc961b9e15a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 18:23:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86f023d1ca2c42e5b0ad7b6b00527ce7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f0f069b20945fcb9ad5fc5e21c991b : SUCCESS in 7m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/95f73e7605484062be480192267d86fb : SUCCESS in 8m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5576d3ec145344ba85e1851f97ab3b26 : SUCCESS in 8m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9cb2a31034e14445bb7a29353140b7d3 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/49039f1a892340f493337067aacc0173 : SUCCESS in 1h 35m 58s","accounts_in_message":[],"_revision_number":6},{"id":"47652701e22416677aa08d2d6c7c69a639c87868","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-13 00:46:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"51aaa57990afe81ce7c22f308e218b0075341f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 00:47:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"783253386b642ce8f34b323cbd45817870898de7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 03:56:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe74aad3767426a854ba8bdc2dba363\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca2da74379b4504a06bfcd13ffb71a9 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4ad4c09ce73243ffa3bb2dcb0fe36246 : SUCCESS in 14m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/68c7b79660554711aabafd67baab92ec : SUCCESS in 11m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/637d95a156ba4f70abf4f3e27d4ce080 : SUCCESS in 3m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bb28541fa16349f5a5bb15b5f3db49b2 : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":6},{"id":"a207de933fa138a48f79a51f96aef7a9fec398ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 03:56:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c81ca83cc20483341bdb706241fce1d138c504e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 03:57:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d428869bd34b98bd3dbb81bce03120\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f58ad69741e4e9f980556fa30e556e2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e93e29c55794244a00da3715feb521dccb4e1d69","revisions":{"88af9d8b5f39d02c6fbf55ef001be106649826cf":{"kind":"REWORK","_number":1,"created":"2026-02-10 22:12:55.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/1"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"subject":"Hide Flavor column on Volumes page","message":"Hide Flavor column on Volumes page\n\nHide the Tags column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/88af9d8b5f39d02c6fbf55ef001be106649826cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/88af9d8b5f39d02c6fbf55ef001be106649826cf"}]},"branch":"refs/heads/master"},"0dfc3d7b7c348429a1fadb0381e364e953e286b3":{"kind":"REWORK","_number":2,"created":"2026-02-10 22:15:41.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/2"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:15:30.000000000","tz":0},"subject":"Hide Flavor column on Volumes page","message":"Hide Flavor column on Volumes page\n\nHide the Tags column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0dfc3d7b7c348429a1fadb0381e364e953e286b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0dfc3d7b7c348429a1fadb0381e364e953e286b3"}]},"branch":"refs/heads/master"},"8d5cf5434a5e32540864a66b534d570952710b3d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-10 22:41:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/976315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/3"}}},"commit":{"parents":[{"commit":"2145076581d80e122a3d2cadf6ba3d4b8c86240d","subject":"Merge \"Fix text wrapping in DetailCard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2145076581d80e122a3d2cadf6ba3d4b8c86240d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 22:41:30.000000000","tz":0},"subject":"Hide Flavor column on Instances page","message":"Hide Flavor column on Instances page\n\nHide the Flavor column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d5cf5434a5e32540864a66b534d570952710b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d5cf5434a5e32540864a66b534d570952710b3d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ea9b3201690eecc968867bb6855acb5dc1be2b94":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-11 09:38:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/4"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-11 09:38:02.000000000","tz":0},"subject":"Hide Flavor column on Instances page","message":"Hide Flavor column on Instances page\n\nHide the Flavor column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea9b3201690eecc968867bb6855acb5dc1be2b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea9b3201690eecc968867bb6855acb5dc1be2b94"}]},"branch":"refs/heads/master","description":"Rebase"},"c75b9349da4f61b85af5e5622305974bfebd2ace":{"kind":"REWORK","_number":5,"created":"2026-02-11 12:53:51.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/5"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-11 12:53:48.000000000","tz":0},"subject":"Hide Flavor column on Instances page","message":"Hide Flavor column on Instances page\n\nHide the Flavor column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c75b9349da4f61b85af5e5622305974bfebd2ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c75b9349da4f61b85af5e5622305974bfebd2ace"}]},"branch":"refs/heads/master"},"e93e29c55794244a00da3715feb521dccb4e1d69":{"kind":"REWORK","_number":6,"created":"2026-02-12 14:07:54.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976315/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976315/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976315/6"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-10 22:12:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 14:07:37.000000000","tz":0},"subject":"Hide Flavor column on Instances page","message":"Hide Flavor column on Instances page\n\nHide the Flavor column to reduce visual overload and improve readability\non the Instances page from Administrator View.\n\nChange-Id: If202a4792ed4272257101d30137befada9c26479\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e93e29c55794244a00da3715feb521dccb4e1d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e93e29c55794244a00da3715feb521dccb4e1d69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
