)]}'
{"id":"openstack%2Fskyline-console~976600","triplet_id":"openstack%2Fskyline-console~master~I55ccd227930793a718ef298a2aff4693774b607a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-13 07:57:46.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-13 00:46:15.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-12 14:14:34.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I55ccd227930793a718ef298a2aff4693774b607a","subject":"Replace Protection Field Name","status":"MERGED","created":"2026-02-12 11:35:27.000000000","updated":"2026-02-13 10:01:09.000000000","submitted":"2026-02-13 10:00:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976600","meta_rev_id":"607b712a9ed6fdd4d10b1a968b9628c2995c6eaa","_number":976600,"virtual_id_number":976600,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 10:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-13 00:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-12 14:14:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-13 00:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 11:56:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-12 11:56:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-12 12:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f166101e84669a6df1cc78d0de3410f18cdee23f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 11:35:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c3e9ba76adad70ca01333cec6925dc7d7183b6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 11:36:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c0510cf79841bd7636cb8232ed6604fdc419e3d8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 11:36:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"238e75f78565e332612c56e3c1d924e3ec59af7e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 11:56:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b4b84799f282661999e4c48d52d4927297d802b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:50:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e3a19e5c73d4d2084c303dec7dbd50e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0232334e03cf4728b3f5588dcf2f62eb : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9a7057342b3247b1bb6c9b2fed84c61e : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1513854a85f7466bafff20b0cb577666 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4ff99ed362ba4ef898d953b95ba763fd : SUCCESS in 7m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/80e924e4cfb148c49db0a2892fd194dd : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"083c6bee8f5ab801c30574d2b5666af4a696d812","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 14:03:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1635c19fe1964f3dc4250adb01a259208bf2153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 14:13:41.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"aab69794a90643622dddbb3fbc9d510320c524d2","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-12 14:13:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f8e07666c09162c43494f24a81f2f41bc98c9ac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 14:14:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57f24244007234238c801b0075dd8a80a55cc1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:48:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b77ff4c1eb4af9846ffd05fe4d4281\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fba8a0451e425bbce4fcdc89b233f2 : SUCCESS in 3m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/672d98a3d27b46f6883ccd81c8523db4 : SUCCESS in 11m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bbe2d4b9207142d9a562a9f5d389fcfa : SUCCESS in 8m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/655291fa087f4b3b9bb78c4f9b459c72 : SUCCESS in 8m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/68d138389a1848ed90719124c08d365f : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":3},{"id":"dbc85b74dfb72a2b63193e079f6f4bc2767c7a45","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-13 00:46:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"488137c1c6ed668ab5296869e535e5e0d252953b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 00:46:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b7f9e26818629162a47e7b1b2a8e8366f94258a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 02:06:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95359bfa29504d24a5b0f81548d34711\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a575b11bd6744469ef9654c4d0bf168 : SUCCESS in 7m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/41484d287c46416295b82f4ccb31fa3f : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/16327a54efff439db9922c9375c70611 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/48557aa384d7424f92fafe56e57a659d : SUCCESS in 3m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/629a92519bc94f0885a150b71196bb78 : FAILURE in 1h 18m 49s","accounts_in_message":[],"_revision_number":3},{"id":"90d61d0d8d0f8246cff66eb9081c9af6fa903db8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-13 07:57:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7870f883b722a0378118d3a3bc9466ec29be5a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 08:49:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46b7abba9d4941f7935efc99c8867b26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8661b2d81176402eb47ee4a417820687 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/96df4b5627644d7897f43b4c54cdc80d : SUCCESS in 7m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/db856290d12b4060a0ea097e7dd2116b : SUCCESS in 4m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/be105356f7ac479dbb3d802e89e007e4 : SUCCESS in 4m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9d010c2df60c47a89f82ea87ec3bf9bf : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":3},{"id":"4bcfef6f57257c2b757ce55882e3a02b17f1f6ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 08:49:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf79d5b6cf449bba8699649243ea63e0972388fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:00:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c82b04ac6ced480a96df327ed3e78db6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc5c180f7904598bfae3ba488263168 : SUCCESS in 7m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6049d0c3047a4eeabe8dcb965e70e05e : SUCCESS in 5m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d7c81e6fb87243f3a48f999b49914528 : SUCCESS in 5m 05s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fe9ca5ae00c94fdb84402984fe766547 : SUCCESS in 5m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7b9484b9367d4b21a9a519ea20896082 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":3},{"id":"2ae8b3be06b4ae91545f2d7977f5eb24c1e91dc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:00:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"607b712a9ed6fdd4d10b1a968b9628c2995c6eaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:01:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989cb3f28a59422c97424016cf669450\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5beb252f819c4750950b31feb5898a9f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a8b187ae36eb2fd791344ab72e24e78ba375b13","revisions":{"5e533d415b76480107f82e2d5344f583bfa8f27a":{"kind":"REWORK","_number":1,"created":"2026-02-12 11:35:27.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/976600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/00/976600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/1"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 11:35:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 11:35:27.000000000","tz":0},"subject":"Replace Protection Field Name","message":"Replace Protection Field Name\n\nReplace \"Protected\" for \"Protection\" field name\nwhile creating Image for more grammar clarity\nand improve UX.\n\nChange-Id: I55ccd227930793a718ef298a2aff4693774b607a\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5e533d415b76480107f82e2d5344f583bfa8f27a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5e533d415b76480107f82e2d5344f583bfa8f27a"}]},"branch":"refs/heads/master"},"78c8f3e885718394e466c02fa4d389fe761fb4d2":{"kind":"REWORK","_number":2,"created":"2026-02-12 11:36:05.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/976600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/00/976600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/2"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 11:35:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 11:36:04.000000000","tz":0},"subject":"Replace Protection Field Name","message":"Replace Protection Field Name\n\nReplace \"Protected\" for \"Protection\" field name\nwhile creating Image for more grammar clarity\nand improve UX.\n\nChange-Id: I55ccd227930793a718ef298a2aff4693774b607a\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78c8f3e885718394e466c02fa4d389fe761fb4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78c8f3e885718394e466c02fa4d389fe761fb4d2"}]},"branch":"refs/heads/master"},"7a8b187ae36eb2fd791344ab72e24e78ba375b13":{"kind":"REWORK","_number":3,"created":"2026-02-12 14:13:41.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/00/976600/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/00/976600/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/00/976600/3"}}},"commit":{"parents":[{"commit":"32dc5fed112d8e7f28a54457b4a8b42ad9cb6255","subject":"Merge \"\"(Sec)\" rename to \"(s)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/32dc5fed112d8e7f28a54457b4a8b42ad9cb6255"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 11:35:27.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-12 14:13:36.000000000","tz":0},"subject":"Replace Protection Field Name","message":"Replace Protection Field Name\n\nReplace \"Protected\" for \"Protection\" field name\nwhile creating Image for more grammar clarity\nand improve UX.\n\nChange-Id: I55ccd227930793a718ef298a2aff4693774b607a\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8b187ae36eb2fd791344ab72e24e78ba375b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8b187ae36eb2fd791344ab72e24e78ba375b13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
