)]}'
{"id":"openstack%2Fskyline-console~976676","triplet_id":"openstack%2Fskyline-console~master~If4de7172d1bdca100c56d343e68b9611a6738fee","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-13 16:44:15.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-13 14:09:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-13 16:44:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4de7172d1bdca100c56d343e68b9611a6738fee","subject":"Remove no need space","status":"MERGED","created":"2026-02-12 18:38:02.000000000","updated":"2026-02-13 16:44:15.000000000","submitted":"2026-02-13 16:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976676","meta_rev_id":"d6614d34c3c3e14e506a562166ce3fa94106e6f6","_number":976676,"virtual_id_number":976676,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 16:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-13 14:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-13 14:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 19:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 12:57:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-13 12:57:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"1fd2a5ee1cacad91a5049e7fc3f387a1f0384fcd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 18:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b54b2c9ef6cdf582af3366479ac547d79e7545b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 18:39:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"70c3d8ff0370d0f46b92ae23ef40e11da54dfdfd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 18:40:01.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b401ab3b7248f3283da56d9edc83a62872315a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:00:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22ce3fa570d14f89b9ff84429859f8ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6cda7903924a5a991a466583732675 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/97186f12e31c4d1ca1cc3a44faecd90a : FAILURE in 9m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a011a0340716475da1b46febc130354b : SUCCESS in 7m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1f0669da88c349bd9c4c04916e1f8766 : SUCCESS in 7m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/980d2b1533824c509ad64822733325c4 : FAILURE in 10m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9acc62c13cef1b7a042747631eaf3c3e426f43bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:12:58.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"81cfd68b052d3a408f33cab103fdf99f1cbbde1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:46:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b24eee1bb384461ca016674e9bc7c22e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b4f55ac1c34f7b8f9f6668e52d2841 : SUCCESS in 6m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/69be9b4f38464df3a8e5bb426d712f11 : FAILURE in 5m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/af94ab9ff98e456c820570623a312f2b : SUCCESS in 8m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2edaa37eee5f4b48bccfaa5a854b090b : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fec94b0da10645c0a74ae8d4cdd4b151 : FAILURE in 27m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8d3bd013f877b2ed9877dd1ad989ae439f7bebbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:49:34.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b4a5a18a2cfef25d8b6661a563a3a3b75f6dc3ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:26:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bedaaf87f79f4ffa9674e2791226a087\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2156d514f74fc0aef1688f4d5039da : SUCCESS in 3m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e505489949944fba8c20949a63876898 : FAILURE in 11m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4728feecbc924239bae23fe2093974c7 : SUCCESS in 8m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/504a4be0221442228013a62b02117bcf : SUCCESS in 4m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a4947d8f31d5463a8fc1acb3d936d4f1 : FAILURE in 31m 16s","accounts_in_message":[],"_revision_number":4},{"id":"22cc57964b21e5ed9099270eb3817fef58a5349e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 10:29:52.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"31d4d1de461020ef365051924dcbd707ae3149b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 11:08:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd74924edb14f1482c0c58df0dbbdc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e092effab5824c0ea1ec52658943c6af : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/56aafb83b7fa42c88a5c949c52f8b8ff : SUCCESS in 4m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/96c4a33c71cc451f86d0df33add7de5b : SUCCESS in 5m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/37ce8d35f2794f83ad39e2d15baa9e17 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1a69a2cecadd4cf697ac307efe119675 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":5},{"id":"a26c822e6405360a4a6039f12868360e6bd018d5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-13 14:09:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e34f8ae40341a0c64b05f4522049fc70db525919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:10:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d11d22d61830f2c210debe26f0dfd23d58d226a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:44:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99528b705b3444f5b358dc14750a9d31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a018941a934becb7a825365fb7d04e : SUCCESS in 6m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a1b764da4e3b4cf7a6208439e9793560 : SUCCESS in 5m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8c1adcb71c9b4ca5969f8f0c2655d9db : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ab60351a741646289b7023fa09af4027 : SUCCESS in 3m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b8c4002c744a4accb04208faa39c92fa : SUCCESS in 1h 15m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d6614d34c3c3e14e506a562166ce3fa94106e6f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f7ebdf791193216eb10f888fa10765cc3d667bfd","revisions":{"03c46af53479d294d77a532a27759dd29bf71365":{"kind":"REWORK","_number":1,"created":"2026-02-12 18:38:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/976676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/1"}}},"commit":{"parents":[{"commit":"a8a43ea0c8712946dbb88afacef10edd85a7b6b9","subject":"Merge \"Add Red Hat and Oracle Linux image types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8a43ea0c8712946dbb88afacef10edd85a7b6b9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If4de7172d1bdca100c56d343e68b9611a6738fee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03c46af53479d294d77a532a27759dd29bf71365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03c46af53479d294d77a532a27759dd29bf71365"}]},"branch":"refs/heads/master"},"540bc8deee380f8fcfaee27d452030f886ed81ab":{"kind":"REWORK","_number":2,"created":"2026-02-12 18:39:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/976676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/2"}}},"commit":{"parents":[{"commit":"a8a43ea0c8712946dbb88afacef10edd85a7b6b9","subject":"Merge \"Add Red Hat and Oracle Linux image types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8a43ea0c8712946dbb88afacef10edd85a7b6b9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:39:56.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If4de7172d1bdca100c56d343e68b9611a6738fee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/540bc8deee380f8fcfaee27d452030f886ed81ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/540bc8deee380f8fcfaee27d452030f886ed81ab"}]},"branch":"refs/heads/master"},"c3684855724c2215caf679385b0a7530fb541bea":{"kind":"REWORK","_number":3,"created":"2026-02-13 09:12:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/976676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/3"}}},"commit":{"parents":[{"commit":"a8a43ea0c8712946dbb88afacef10edd85a7b6b9","subject":"Merge \"Add Red Hat and Oracle Linux image types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8a43ea0c8712946dbb88afacef10edd85a7b6b9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:12:57.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If4de7172d1bdca100c56d343e68b9611a6738fee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c3684855724c2215caf679385b0a7530fb541bea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c3684855724c2215caf679385b0a7530fb541bea"}]},"branch":"refs/heads/master"},"df428f44c123a3bccdf9bd7bde72a4cc8b2d2386":{"kind":"REWORK","_number":4,"created":"2026-02-13 09:49:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/976676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/4"}}},"commit":{"parents":[{"commit":"a8a43ea0c8712946dbb88afacef10edd85a7b6b9","subject":"Merge \"Add Red Hat and Oracle Linux image types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8a43ea0c8712946dbb88afacef10edd85a7b6b9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:49:34.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If4de7172d1bdca100c56d343e68b9611a6738fee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df428f44c123a3bccdf9bd7bde72a4cc8b2d2386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df428f44c123a3bccdf9bd7bde72a4cc8b2d2386"}]},"branch":"refs/heads/master"},"f7ebdf791193216eb10f888fa10765cc3d667bfd":{"kind":"REWORK","_number":5,"created":"2026-02-13 10:29:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/976676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/976676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/976676/5"}}},"commit":{"parents":[{"commit":"a8a43ea0c8712946dbb88afacef10edd85a7b6b9","subject":"Merge \"Add Red Hat and Oracle Linux image types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8a43ea0c8712946dbb88afacef10edd85a7b6b9"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-12 18:38:02.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 10:29:52.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If4de7172d1bdca100c56d343e68b9611a6738fee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ebdf791193216eb10f888fa10765cc3d667bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ebdf791193216eb10f888fa10765cc3d667bfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
