)]}'
{"id":"openstack%2Fskyline-console~976915","triplet_id":"openstack%2Fskyline-console~master~I60ef6fc48cc0b94945367336a10edb860e01a4c4","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-18 13:58:55.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-18 12:34:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-18 13:58:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60ef6fc48cc0b94945367336a10edb860e01a4c4","subject":"Update Russian language","status":"MERGED","created":"2026-02-16 11:29:31.000000000","updated":"2026-02-18 13:58:55.000000000","submitted":"2026-02-18 13:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976915","meta_rev_id":"8a0acfee3ad79a2c9b8f4c49015b506c735ac558","_number":976915,"virtual_id_number":976915,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 13:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-02-16 21:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-02-18 12:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-02-18 12:34:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 11:30:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-16 13:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 21:16:27.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"9234783408fe2b8af7a007175af42a38ec5694aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 11:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"593e3f86fa5c3e0a7c36a80d73ae92995645bf3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:08:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2c4ff4130d44f1bb7a74977dd264ab5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/242818fe05ed4545b55cf5e613aafdaf : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c9f8259933e4c13b9a65fe22938f23d : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e680b9cc2fa44ce88d1bfbd6c995ed8 : SUCCESS in 5m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/080e47f6f9584269849f25471e2209f9 : SUCCESS in 11m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d51903be8b5e4106b7777a82f90df47b : SUCCESS in 7m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7d9dba7b9d444220b55f05373c993189 : SUCCESS in 6m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/80c7e760e57746e19315b675af9a61dd : SUCCESS in 1h 21m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/60c966a8db9a43bd8d73761fbb0d8ea6 : SUCCESS in 32m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/152076486caa46dda373b8984c9ff44e : SUCCESS in 28m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/df03452339f64ebeb26a06bb4d1eef19 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9d33634cb8cdbe65c5ebb148feaa1b2d05c7b72d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 16:35:52.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"90b78b9c8ad3441702534d879685bb1dbb860cfe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 16:42:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1694fe079fce51760f94dd3c959b791e9fb18ecd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 16:42:55.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"6a0c2317c05cd676b336a3f2f24bfe61497b7dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 17:05:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf37448475140daa56068a95f8119c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a4e6917e4b4a61b0818bcd3cda62ae : SUCCESS in 8m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/31d010e722cc435eb594dd3fe3ab32c0 : SUCCESS in 14m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e4a73a2d7a724ecba78954231adfde90 : SUCCESS in 7m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/514f57495a40476a89a7db32c9e0e4f1 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":3},{"id":"d2f755377e963d12b8db10ccf745c48dd887ca2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 17:37:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"34047415411dd42dfd1bc23698f62a2f29b2351a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 17:45:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a9ebc6c4c40b071052c70af9ab2792ffb07914da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:06:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab5a879c429452ebacc7a0df270cc77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4676169cbd84779a438b73e4144d60c : SUCCESS in 8m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0db45bad221448bab4d2a003b3f0b79d : SUCCESS in 4m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c171d8cc463547e2be24e69b15f1992a : SUCCESS in 4m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c1b851b3c7d241b9a8a11a1c1e4ecce8 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":5},{"id":"7ae054b220cea4426fb523dd9f7b7ba9d486ca1f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 21:16:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0d971c2ef6a53cd012605dbc6b42dbe1c67f45d0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-18 12:34:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e0c6bab002d84a5f29bd0dcc01c0721cada6d124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 12:34:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7fdc3cc2590c9d4b20731ff5e9c5c11e750c64d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:58:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a9e6fa4f9b47a884aaf828b3bff034\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd307bfecd94eb598c51fe706fdc859 : SUCCESS in 6m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cab7b0b0f9594c1793e904c4a6b9f102 : SUCCESS in 9m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6325bff2bd054531b3252a7c166b9d96 : SUCCESS in 5m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/64f112d98deb4c9eb1c175daf123a01c : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8a0acfee3ad79a2c9b8f4c49015b506c735ac558","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:58:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c9771f7921331a36bbadd3fb575c42f1a0d07431","revisions":{"abf7b27beecbdd059ecf27613086859f8dac4aef":{"kind":"REWORK","_number":1,"created":"2026-02-16 11:29:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/976915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/1"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I60ef6fc48cc0b94945367336a10edb860e01a4c4\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/abf7b27beecbdd059ecf27613086859f8dac4aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/abf7b27beecbdd059ecf27613086859f8dac4aef"}]},"branch":"refs/heads/master"},"f3739f764efcfce47b8f7615b114db7a8ef689c4":{"kind":"REWORK","_number":2,"created":"2026-02-16 16:35:52.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/2"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 16:35:02.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I60ef6fc48cc0b94945367336a10edb860e01a4c4\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3739f764efcfce47b8f7615b114db7a8ef689c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3739f764efcfce47b8f7615b114db7a8ef689c4"}]},"branch":"refs/heads/master"},"55bd10d300970ce65d91ae12d8ce74df2e63aadd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-16 16:42:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/15/976915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/3"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 16:42:52.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I60ef6fc48cc0b94945367336a10edb860e01a4c4\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/55bd10d300970ce65d91ae12d8ce74df2e63aadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/55bd10d300970ce65d91ae12d8ce74df2e63aadd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a04b84f2d6f73c713bbe869a9eadc00ebfe5f242":{"kind":"REWORK","_number":4,"created":"2026-02-16 17:37:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976915/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976915/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/4"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:37:28.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I60ef6fc48cc0b94945367336a10edb860e01a4c4\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a04b84f2d6f73c713bbe869a9eadc00ebfe5f242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a04b84f2d6f73c713bbe869a9eadc00ebfe5f242"}]},"branch":"refs/heads/master"},"c9771f7921331a36bbadd3fb575c42f1a0d07431":{"kind":"REWORK","_number":5,"created":"2026-02-16 17:45:38.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/976915/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/976915/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/976915/5"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-16 11:29:31.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:45:15.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I60ef6fc48cc0b94945367336a10edb860e01a4c4\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c9771f7921331a36bbadd3fb575c42f1a0d07431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c9771f7921331a36bbadd3fb575c42f1a0d07431"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
