)]}'
{"id":"openstack%2Fskyline-console~977016","triplet_id":"openstack%2Fskyline-console~master~I90201facce5e1df7f7240b186dba9b77118c4b4a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-18 13:58:53.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-18 12:34:04.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I90201facce5e1df7f7240b186dba9b77118c4b4a","subject":"Bring  \"Create Host Aggregate\" to one standard","status":"MERGED","created":"2026-02-16 17:35:43.000000000","updated":"2026-02-18 14:00:02.000000000","submitted":"2026-02-18 13:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977016","meta_rev_id":"51f47acce1e0b51e98e5c046f6ddbdab759ae68d","_number":977016,"virtual_id_number":977016,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 13:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-18 12:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-16 18:23:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-18 12:34:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 18:23:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-16 18:23:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-16 19:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa337a7db2fa0140d798e874d720e320788b410f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 17:35:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"191d386283ee52e5a4656dede141d75143c3f432","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 17:36:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"29ce56762ed8af4de8e3fafe89113939343e0785","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-16 17:36:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7a7f32312f557a80f2feceedfec067dec6eb7e74","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 18:23:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c2a005e1aeaa359ec427253ba77745e23bcc0cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73197f3d783d4547926ef549ade4ea25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568fc54ae87f43b5bf900d0ebed36eac : SUCCESS in 3m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/965da37e94cd4ca384e2f11fc76fe762 : SUCCESS in 11m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6411cb885e424066b146dc88299a0d7c : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a568ef34d5664840b6c1b0779ab3a642 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/da71f66e9e4a44a3868d618c63757043 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":2},{"id":"274ad84184ac110bcee47182c4e4dd48d1505404","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-18 12:34:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"039527057ced5da2289a5547d838f3151059f52f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-18 12:34:10.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b7fa3ace3f82241a424e3520d1b5f9bade1053b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 12:34:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4fbfef7b5249cd764520b60dc7ac2cf67355a32a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:58:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1354e8641548493d81d9bd8c1f791429\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e96f0c504464d72989ca28505cca91c : SUCCESS in 8m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a39cd72abae74d63842dca898d2babef : SUCCESS in 7m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a0768d734dd94c2983d32f70d58730d9 : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/82e06ed0cc784df88c28546048984479 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/eda5dbfddd90436f90ccd8c1107be48a : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1b781d15fdca11ee7953e5d4abcbbea174dda4b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 13:58:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51f47acce1e0b51e98e5c046f6ddbdab759ae68d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 14:00:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17958f1a113f4778ab218328fc40f500\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2251b10771c84ebcb0297a9248e9e42d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2a7fff158c6f93c4c066c0a4deecb925180d800","revisions":{"6eee9d52104f4136a270714e1f2786236dbbf790":{"kind":"REWORK","_number":1,"created":"2026-02-16 17:35:43.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/16/977016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/977016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/1"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:35:43.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:35:43.000000000","tz":0},"subject":"Bring  \"Create Host Aggregate\" to one standard","message":"Bring  \"Create Host Aggregate\" to one standard\n\nChange-Id: I90201facce5e1df7f7240b186dba9b77118c4b4a\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6eee9d52104f4136a270714e1f2786236dbbf790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6eee9d52104f4136a270714e1f2786236dbbf790"}]},"branch":"refs/heads/master"},"a2a7fff158c6f93c4c066c0a4deecb925180d800":{"kind":"REWORK","_number":2,"created":"2026-02-16 17:36:35.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/16/977016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/977016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/977016/2"}}},"commit":{"parents":[{"commit":"7ee3be22cf56d6044e751822988e2df404439240","subject":"Merge \"feat: add Network Quick Start functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee3be22cf56d6044e751822988e2df404439240"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:35:43.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-16 17:36:27.000000000","tz":0},"subject":"Bring  \"Create Host Aggregate\" to one standard","message":"Bring  \"Create Host Aggregate\" to one standard\n\nChange-Id: I90201facce5e1df7f7240b186dba9b77118c4b4a\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2a7fff158c6f93c4c066c0a4deecb925180d800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2a7fff158c6f93c4c066c0a4deecb925180d800"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
