)]}'
{"id":"openstack%2Fskyline-console~977324","triplet_id":"openstack%2Fskyline-console~master~Ia934d5445081ae3fff591d44428fce138fd1fd17","project":"openstack/skyline-console","branch":"master","topic":"any_availability_zone","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-24 09:44:21.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-24 01:44:42.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"5575":{"account":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"last_update":"2026-02-24 07:04:29.000000000","reason":"\u003cGERRIT_ACCOUNT_5575\u003e replied on the change","reason_account":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"}},"38847":{"account":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"last_update":"2026-02-24 09:44:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia934d5445081ae3fff591d44428fce138fd1fd17","subject":"Allow to specify any availability zone on instance creation","status":"MERGED","created":"2026-02-19 14:35:09.000000000","updated":"2026-02-24 09:45:17.000000000","submitted":"2026-02-24 09:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977324-any_availability_zone","meta_rev_id":"d3ec1475bd2919480dcf4eaa6a076cbdb559e027","_number":977324,"virtual_id_number":977324,"owner":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 09:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-24 01:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2026-02-24 07:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-22 13:08:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-24 01:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 14:43:50.000000000","updated_by":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"reviewer":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"state":"REVIEWER"},{"updated":"2026-02-19 16:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-22 13:08:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-22 13:08:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-22 13:08:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"e75f1b96b234a2b4be2106d5b39e4a41379992ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-02-19 14:35:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a1db1738def6f5cc857a451ca084bb5496fb138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-02-19 14:37:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0cdce6505a499facba9a7a30a0c119539f8d6056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:06:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db0756f863843318862ca1731fd4581\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f0630b503c4d2bb27aa63da2bfe939 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bf28e99f10ec4d98be6f492316382f04 : SUCCESS in 4m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b6b9d2d2841942c8922d5af07635d2aa : SUCCESS in 7m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f676228beb144178af2296129168976c : SUCCESS in 5m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bf0ecf2b731144c9b257a68d29ba37d7 : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4e85c3d44b1ad0263c1beb600acee0e59fcb4be6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 13:08:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ebb7ef3d0ca96ca235cd20e33cabdfa0a1c5f89a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-24 01:44:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b54e18638589274de6d153611b702444ccf271c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 01:45:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40a68a158874997b3fa42602e376f76ba2463eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 02:44:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fbc49b9b90e42b6a3a7fb9bb961ec3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71842139dd9047aa8e2abfd822929fde : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2bc53c24f5fe427e93597c29d28488b7 : FAILURE in 2m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a88840f7ad8c42e39f518dbb6ce53699 : SUCCESS in 3m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/13c359276edd4f5185fa71957c24b26a : SUCCESS in 9m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e2f98a574dc94997bd0892f69a068c47 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"9c3404b4b672aa31336391af6c5b54599b1585a3","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2026-02-24 07:04:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d5bde4c02db604e59bebbebee59a30a0b783997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:27:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0bb8eedc464d8097980bff04f9fa48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0584c11219b4d25bc14af9c460f9277 : SUCCESS in 5m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1bcec533985646ce83d54838ca642fc3 : SUCCESS in 8m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0ada0741a4c946c68b27e67a2791d73e : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0242dd4e321c4960be503681cbd69de3 : SUCCESS in 3m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4c94249d445d44aca92a94d552e9d4a1 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a62c235a48ede8084fedafa0b94bc595f88e4617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:28:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"671e9678421228c16c79a68299ce751dacfa50f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:44:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7511de7fe74e90bb06baa2be13a9f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc946f59e8f840b6a71a283c71f642e9 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5d0e5515e4e940a8b73a300aed2d69b9 : SUCCESS in 5m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4da56188f3bd495fa70147adb8017956 : SUCCESS in 6m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f9ed20f11c4b499a8349b96378d65e07 : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cbb22d8cbce44b629e45de234691857c : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c28b0aa340b0c8dc2923568b6413e62165c13a4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:44:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3ec1475bd2919480dcf4eaa6a076cbdb559e027","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:45:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea521d3d4e8b4f2f9ce41f187f30e8a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f1e6371a9742deb6936cef2d25f168 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"38e11b5a0194b7530cdca6f97e5461777b80e3be","revisions":{"bdd7f10d508d6e95c27ae718d05c680934dd329a":{"kind":"REWORK","_number":1,"created":"2026-02-19 14:35:09.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/24/977324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/24/977324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/1"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-02-19 14:26:57.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-02-19 14:36:27.000000000","tz":60},"subject":"Allow to specify any availability zone on instance creation","message":"Allow to specify any availability zone on instance creation\n\nThis enables an option \"Any Availability Zone\" in the stepcreate panel.\nThen on submission, the argument to server create will not be sent,\nleaving to the nova backend to select the most desirable availability zone.\n\nCloses-Bug: #2067812\nChange-Id: Ia934d5445081ae3fff591d44428fce138fd1fd17\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bdd7f10d508d6e95c27ae718d05c680934dd329a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bdd7f10d508d6e95c27ae718d05c680934dd329a"}]},"branch":"refs/heads/master"},"38e11b5a0194b7530cdca6f97e5461777b80e3be":{"kind":"REWORK","_number":2,"created":"2026-02-19 14:37:56.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/24/977324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/24/977324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/24/977324/2"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-02-19 14:26:57.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-02-19 14:38:45.000000000","tz":60},"subject":"Allow to specify any availability zone on instance creation","message":"Allow to specify any availability zone on instance creation\n\nThis enables an option \"Any Availability Zone\" in the stepcreate panel.\nThen on submission, the argument to server create will not be sent,\nleaving to the nova backend to select the most desirable availability zone.\n\nCloses-Bug: #2067812\nChange-Id: Ia934d5445081ae3fff591d44428fce138fd1fd17\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38e11b5a0194b7530cdca6f97e5461777b80e3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/38e11b5a0194b7530cdca6f97e5461777b80e3be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
