)]}'
{"id":"openstack%2Fskyline-console~977418","triplet_id":"openstack%2Fskyline-console~master~I8be3e49e4eed0c47747c413597a775b9b4722e12","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-24 02:53:16.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-24 01:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2026-02-24 02:53:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8be3e49e4eed0c47747c413597a775b9b4722e12","subject":"feat: Show console in rescue mode.","status":"MERGED","created":"2026-02-20 07:10:33.000000000","updated":"2026-02-24 02:54:51.000000000","submitted":"2026-02-24 02:53:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977418","meta_rev_id":"4cfcb85f603074528fac2b894be52487197b6f0d","_number":977418,"virtual_id_number":977418,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 02:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-24 01:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-22 13:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-24 01:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-20 07:11:26.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-20 08:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-22 13:08:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-22 13:08:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"88d86e284dbd22d4c9ea13332f821183cc0afa13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-02-20 07:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57c0cc1a4edbd2529fdef94124a72b863637ad8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-02-20 07:10:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"26a6ae682c2ce1d9e39478a56fe19da9589d2020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 08:29:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d90f9f46e0ca48f7b0451eedec355a4d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/912b152373ff43c9a32f5e74b3304d34 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f4f1f5a2184618904b4d8b6a83361b : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e197a08b47c342059ebdd934f218b7f5 : SUCCESS in 9m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0ba2ca8ebc774cc09498c742a9ebc0b0 : SUCCESS in 6m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/363c861b2fdd41a0935044f79649003b : SUCCESS in 4m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4094a24d439c4911819fb2c74df21925 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ef87b9fd9e5c59c1f7fc2f09c22badf75330c741","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 13:08:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d92eea859890060ebdfd11c78f5250c57fcbee93","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-24 01:45:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c42344d186654ac7827f71eb3a60289b7bb1a7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 01:45:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b000d62428ba82de33f0f9260cf8d9a6cb10fb3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 02:53:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb743af62ee47f2a5cd702f078008dd\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b5413bb6db45a6892c7735abdc3fba : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8660dbad9f96413298980d8795a8334e : SUCCESS in 5m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/27e8755f8e934beb89b50ad00409d826 : SUCCESS in 7m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a90b0b5369f448a9a630e42bf475a882 : SUCCESS in 6m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/244a1431f72041bd94e03a33b26349e9 : SUCCESS in 3m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fb28b78276a24cfc9785a3168531509b : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":2},{"id":"dc3b022609c6ffd733f4da5f97912b571a2af592","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 02:53:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4cfcb85f603074528fac2b894be52487197b6f0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 02:54:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fadbf9bfb614cc09148ac740ea706bc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/097183b8f8034084a2a3de294b1fc1a6 : SUCCESS in 1m 15s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7268ba305b44ec58c3dadb67eb6a957 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"615ef75a5b475c9bd012095e669cc4d94a5cf5e5","revisions":{"022ca4b624be1214699ee11e25871d7a8f7686be":{"kind":"REWORK","_number":1,"created":"2026-02-20 07:10:33.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/18/977418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/977418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/1"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-02-20 07:03:30.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-02-20 07:10:32.000000000","tz":0},"subject":"feat Show console in rescue mode.","message":"feat Show console in rescue mode.\n\nChange-Id: I8be3e49e4eed0c47747c413597a775b9b4722e12\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/022ca4b624be1214699ee11e25871d7a8f7686be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/022ca4b624be1214699ee11e25871d7a8f7686be"}]},"branch":"refs/heads/master"},"615ef75a5b475c9bd012095e669cc4d94a5cf5e5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-20 07:10:53.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/18/977418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/977418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/977418/2"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-02-20 07:03:30.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2026-02-20 07:10:53.000000000","tz":0},"subject":"feat: Show console in rescue mode.","message":"feat: Show console in rescue mode.\n\nChange-Id: I8be3e49e4eed0c47747c413597a775b9b4722e12\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/615ef75a5b475c9bd012095e669cc4d94a5cf5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/615ef75a5b475c9bd012095e669cc4d94a5cf5e5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
