)]}'
{"id":"openstack%2Fskyline-console~977580","triplet_id":"openstack%2Fskyline-console~master~I047fe8cafd659fbb55d0b1204a5b975e686e0b4a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-26 19:24:49.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-27 01:51:06.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-27 02:03:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I047fe8cafd659fbb55d0b1204a5b975e686e0b4a","subject":"Update Russian language","status":"MERGED","created":"2026-02-22 18:35:04.000000000","updated":"2026-02-27 02:18:42.000000000","submitted":"2026-02-27 02:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977580","meta_rev_id":"968c6db288fb212b790474400c1e79ddf9e55118","_number":977580,"virtual_id_number":977580,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-27 02:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-02-26 19:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-02-27 01:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-02-27 01:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-22 20:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 19:08:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-26 19:22:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"bd40dc991a9f440084864dc1550f3e2a741f64d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 18:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c465b009c5637bc769c171289a4c830000dbc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 20:09:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c6d6ec068d4a10a6c4fc3b9b0407ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bf18d355b47e4c5293c5ad2ee1355114 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/495b8f29ee594902a53d9bd323a2b4d6 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46e24d7bfbb416f93e92108fe2f65ef : SUCCESS in 8m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/06e3f5164b3248cfb1d8445ee1076438 : SUCCESS in 10m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/369cb1b6bd1647e4abb7d200c32eea00 : SUCCESS in 3m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1e954275a8274ceb947d6ef7aeceecf4 : SUCCESS in 3m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f11e7225cdbc4d0a933f05e80da2ab8b : SUCCESS in 1h 28m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1738af2977074ea3b66114870027dbdb : SUCCESS in 31m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/084394cb78a9481abc91362e664e5a9b : SUCCESS in 37m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0a259619a2c04401bcad6864f50f2967 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5c15f44f3b8525f48316d9777a9afd1988ce8cfb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-26 18:27:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"44b4cb728ae412b1dc8d2d63519bcfb0a031e71c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:44:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37123c7cee346d08139e2fec9662450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b270c18b6fa4bd7b1b58470803d15ac : SUCCESS in 4m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/918ae7ef34ca46869f793f3d3662dfe1 : SUCCESS in 11m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d25610390cac40ff8bd1d3f5a9c24428 : SUCCESS in 6m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6186487269334de6b68a86d4eacd2b20 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"76a6a0fe306b419f9fbdb681e826c06c79868f77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 19:02:17.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e5c36ed25172bbb2b95daf8ba1dd51130bc42af3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 19:03:07.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"abc337319d8b65a9e0a4c04faaf408accbf5e672","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 19:06:09.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"286a404f0fb82660a876b393c21c84ce0c9af904","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 19:07:48.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3bcaa32771c7830cb7abcff849256a012fe636a4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 19:08:00.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"5acdb182b44f11f8db7d6c713930693c2e8e02fa","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-26 19:24:49.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"22e40868b2f7774fe2415c8e8058d620c1b11038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:28:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/347172d1ee73424c9ed90a9c3c48bf13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe704f9bf1e438293eb40503cbe2c56 : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7fd5593cc1e74110818de8c3c7886bc8 : SUCCESS in 11m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/810bb872453d41d785273f5ca9bd5ae9 : SUCCESS in 9m 55s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cb083504a70141b99c4c28a35fd10156 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":6},{"id":"f7d57ab263dd8640758eb4c8648c6ec59854f02a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-27 01:51:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a2f43c6f67a526954410296ade9c9279da7fdd43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 01:51:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0c99d31345c95b6de624494e5bb4455f2d8cd56d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 02:03:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a28b4e9ca1aa4c10a1f02621924fd094\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425827d3022b4ab3b5eb0257684e3c38 : SUCCESS in 5m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/62c64d565ed044b6b2bedf4095c3b495 : SUCCESS in 6m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6a35cd70f80c44e280021f36d486320f : SUCCESS in 7m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ceecce9761004687a2e2482f9c1d0af8 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"dfd5092919993a07182ee84faf8df8febb110866","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 02:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"968c6db288fb212b790474400c1e79ddf9e55118","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 02:18:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bee8741c3b4a40bb60f52e406c59b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cb002b9e29431bb87950d1ffd9be52 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a87e666ec0175df0c92454056e8c740777e24706","revisions":{"0a85fcec396903e7b65baf97b8f14937876cf09e":{"kind":"REWORK","_number":1,"created":"2026-02-22 18:35:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/977580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/1"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a85fcec396903e7b65baf97b8f14937876cf09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0a85fcec396903e7b65baf97b8f14937876cf09e"}]},"branch":"refs/heads/master"},"89578e3dea5871256794cf66d3cca83d20139618":{"kind":"REWORK","_number":2,"created":"2026-02-26 18:27:54.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/80/977580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/2"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-26 18:25:51.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/89578e3dea5871256794cf66d3cca83d20139618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/89578e3dea5871256794cf66d3cca83d20139618"}]},"branch":"refs/heads/master"},"c373f5bb687509faa691f4b57da27182d8871c6c":{"kind":"REWORK","_number":3,"created":"2026-02-26 19:02:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/977580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/3"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-26 19:02:17.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c373f5bb687509faa691f4b57da27182d8871c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c373f5bb687509faa691f4b57da27182d8871c6c"}]},"branch":"refs/heads/master"},"e796117c8bc76ca61875f51fe36208e031b9f1af":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-26 19:03:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/977580/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/4"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-26 19:03:07.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e796117c8bc76ca61875f51fe36208e031b9f1af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e796117c8bc76ca61875f51fe36208e031b9f1af"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a5d4fc6e2dd3a0b7394d402f52247eedb23d9a57":{"kind":"REWORK","_number":5,"created":"2026-02-26 19:06:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/977580/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/5"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-26 19:06:08.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a5d4fc6e2dd3a0b7394d402f52247eedb23d9a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a5d4fc6e2dd3a0b7394d402f52247eedb23d9a57"}]},"branch":"refs/heads/master"},"a87e666ec0175df0c92454056e8c740777e24706":{"kind":"REWORK","_number":6,"created":"2026-02-26 19:07:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/977580/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/977580/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/977580/6"}}},"commit":{"parents":[{"commit":"8e5ec482338703589a46ef959832c51ea283e75f","subject":"Merge \"Fix spelling for new password const\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e5ec482338703589a46ef959832c51ea283e75f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 18:35:04.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-26 19:07:48.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I047fe8cafd659fbb55d0b1204a5b975e686e0b4a\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a87e666ec0175df0c92454056e8c740777e24706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a87e666ec0175df0c92454056e8c740777e24706"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
