)]}'
{"id":"openstack%2Fskyline-console~980897","triplet_id":"openstack%2Fskyline-console~stable%2F2026.1~I26f6db2c2796e9b2e93e87cd93d3289015f9fac1","project":"openstack/skyline-console","branch":"stable/2026.1","topic":"create-2026.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-03-30 01:03:42.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I26f6db2c2796e9b2e93e87cd93d3289015f9fac1","subject":"Update .gitreview for stable/2026.1","status":"MERGED","created":"2026-03-17 11:03:38.000000000","updated":"2026-03-30 07:02:22.000000000","submitted":"2026-03-30 07:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980897-create-2026.1","meta_rev_id":"dd45e5c48da8edb7e3ab556111693000cc54966b","_number":980897,"virtual_id_number":980897,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 07:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-03-30 01:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 01:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-03-30 01:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 11:19:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-21 19:33:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-21 19:33:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"3c27f49953b2f434983543c4f4aceb60a9091d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-17 11:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66e5ca64f339ff492884d07568e12d0f357e2453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:19:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c166bc097514988bdabb092687b23f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c79ce6be93b44dfbad1786575540864 : SUCCESS in 6m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1197cf1ea72c42aabbfc7b2829175824 : SUCCESS in 7m 10s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1619dfd2c74d4bc0917f7d3edb155fd0 : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e05d91b99d71421e934a621d12bc9002 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4cf25782a3d9aa05df384cc09a952742e9a801f6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-21 19:33:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b56ffde962621288d7f45b35f9718f04706c2594","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-23 06:35:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cd753463ce198f0e195b532e73e2c6b3a44ffd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 06:35:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b5c0f6443295dd7c366dcfd690b4cb918db8496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:16:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7aab14daf74d12999932ac95feecfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7932d2dd0504a81a6af2e58c55c8040 : SUCCESS in 6m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ed509cfa2f414040a84c6f1a9484cd03 : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b43a00380ad3433d916fa4bc25127eef : SUCCESS in 3m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cfa76a566bdb47bba7f5e9cce208c592 : SUCCESS in 3m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6959955d4e204df19e3303fb1db45223 : FAILURE in 26m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9dfa923d19c947b3add9e42c40541270 : SUCCESS in 33m 07s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/48f510086be646479992b9850438f27e : SUCCESS in 40m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fd4410d35d69471d9e123bd3fd066f83 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d948f1fcff6e27bdcc376848328bdea8d9be6989","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 20:24:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a38b21babd3457e5405e1315ce06f36dd8fe4e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 21:30:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6ce41cf38b84173b5a58cef672e9970\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ccc710c411459486a9613e016cf669 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/078d00074a6943b7b3e4708e418792cc : SUCCESS in 5m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5a1a2518eaa2430c8eb1748943642300 : SUCCESS in 7m 30s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9d52a5e5aef64e979217f7c21d585de4 : SUCCESS in 8m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3d8933f48df047be8b4321cc221263c3 : FAILURE in 51m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/81a03a6338a547a29077e51e4d0ce7d0 : SUCCESS in 1h 00m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4fc831014acb43ea94b9566ab9a484f3 : SUCCESS in 53m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/726aeed47ede41648f8430533f260abc : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":1},{"id":"23577e518910dba7fd80b08e17a1a8b731eb7a11","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 02:16:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"409732169c963508c9224be82b5eddba6900e4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 04:49:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a80783c71d248c2ba84a70c030d9e88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1a7ee870924d70888db3917b32e7a3 : SUCCESS in 6m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3605318edf7a4a7caab14818a69d62cf : SUCCESS in 4m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/409ec84606f84301811764f2fd5bba4d : SUCCESS in 3m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e87443af51064a6c890770c4b02e964e : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cefcd394e8774a29902c86a7daec2589 : FAILURE in 38m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/43597de50bb543108b683ec3858b8fb8 : SUCCESS in 1h 52m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b352cb7464924666a9fc61b86492576d : SUCCESS in 53m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ccd6c90f2bbe41348f568e1436357450 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cdabea9b7a7e2babf192930fbfc763c08ddde18b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-28 16:29:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3cef4ab03f468817cb2d8613c5477fc4f9e0f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 17:15:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3ace97b1eb5478d9fc1a6b81688ea11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e72fdd8708d4d92ae1d5001d8af7d66 : SUCCESS in 8m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d9d88ee40b34433da79dd5c5f802329a : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7fede48826514c8d87db33a010259950 : SUCCESS in 8m 05s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/93edeb3293ca4463a75a1ebd09c98212 : SUCCESS in 4m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4136fe35a6954e3b9e23669d88917d17 : FAILURE in 33m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2b432175a32d4d4088f2ce4f89d1944f : SUCCESS in 34m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/79aced398034466fb683641ae5a0af1b : SUCCESS in 37m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d107624ddd1d49a089300d3454dffece : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8c257ea837584e54c1aeabc4f070137d05c0d283","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 03:37:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e65ef99301c33afa6c4fcd63deeb65b67426638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 04:58:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c88775190d7d4cd7b026841135b4df3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8df04bc0744fa3bc676f80c2a2cb29 : SUCCESS in 3m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0b0b5af8913d4a6eb56e6887d2b922d1 : SUCCESS in 4m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dfdd520468e541fdbd84999f469abb89 : SUCCESS in 5m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/87d37ab3e7524e8f92436cd834948b63 : SUCCESS in 3m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b7845d96a3c64297b8a84af48a9a8d50 : FAILURE in 45m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e947074105ce47e1a62cd9f463a69e60 : SUCCESS in 1h 19m 22s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/488d54970cf141e69f6d400133903ee5 : SUCCESS in 38m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d7418074fdf546e7b6ddc27533d7aed9 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"38ef87c85f118439a85f828580e363bd88006c65","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 09:02:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cb8670116cba40834742b48d54f4181e083a6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 10:12:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c6085abb05442ef9a38389673a54faa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d15db6ddd44c1fb8d60d9f876e8adf : SUCCESS in 3m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/86e902b4acd84b83809a69b7385059c8 : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/35c4262d4502429892a93b4aa4c0c038 : SUCCESS in 7m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e57ad354af004b9d9f7ba4c10c8d44f1 : SUCCESS in 7m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/201b7a46292e44f18983dd63dcaeb8de : FAILURE in 32m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b68a5914a2d0499d9da5544a4975e825 : SUCCESS in 1h 03m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cf57915e816a4a3798d4a08570360394 : SUCCESS in 53m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f1f6d6e612ce400f86fc8ebc79c03419 : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a192177b9cddb16f47dea9436fb1b907b74ee23a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 21:14:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc6cbcc168b5524adeeb0f4d86c6931e0d113560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 22:43:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf791df68f44b908e4af63a57728253\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb51c7f4ed8496db38370f049d19332 : SUCCESS in 6m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b0b552d9bbb348f993650915aa856de6 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5dda40e1b5dc460a9d1f6bca0d24ec0c : SUCCESS in 5m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e4d1094eaad249a9bdd75c77730f296e : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1240ea88caed43488d5bb26dd0933ad3 : SUCCESS in 1h 24m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0cbf0d8e0ebd4c31970628a9d3227d3d : SUCCESS in 1h 00m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f18ffccd792345eca92febbb9d2f0a6e : SUCCESS in 51m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/106fb187045448aea0047ed626ac66c0 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"54c47553e735b2d0ae218e088ceaf1620cf35bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 22:43:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d46b72d0c0b8bc8b87b3a91508d37557e2213ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 00:11:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e1d1f7820b24be1b8e6bfc067546b58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1410171cba4930ad26d304a92bf309 : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ad8230919deb4091976992e31243bd15 : SUCCESS in 6m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ac3730d3e2df48e19fbf8624d42fdd63 : SUCCESS in 7m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0290849597ef4d709850c61fe4c316f8 : SUCCESS in 4m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/43bac71d79da4147a068fd4eb1caa3cf : SUCCESS in 1h 22m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f69f0d549c9447b4af5795facd1f21e4 : SUCCESS in 43m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ba997e04a00e404faad470357b9680b1 : FAILURE in 43m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f718ad88c27347bf95adc198dcbf3f28 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a13eb3f5fb02341e93c45a09e2177ba6d5c30db9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-30 00:27:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70f46f57e1414473b9e5fbc41ba0c86c79a3f34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 01:03:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"f9a3835c27ba5c96085a55cd874dab20e1a11105","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 01:03:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e37dc979b9aaf7a6a17b3b42a3efeb3493907e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 02:32:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93faf0b62194e33b4b3073f3cdaf088\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837a984e243c4ab9b1f9c1684e1fbf2e : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ea94f878fd2e4ceab8a2cc8e31b6e181 : SUCCESS in 10m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fad2a67f7d4143ad9117eca459a48e05 : SUCCESS in 7m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/98d57c0d8bc34bfea1a4c9a8bc7a6508 : SUCCESS in 3m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/102ee20d7fff40b783e3fd4cd4f784d6 : SUCCESS in 1h 23m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/75b8d743443d4a4384418224a6e3c81a : SUCCESS in 32m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a59a8de1ad1d43a18f2f97c5a46d5515 : SUCCESS in 55m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f62a6b4e51e54e6992f13f8516d22aec : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"07e60f3910b5341c44125c8eaa90bfa06eb60f26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 02:32:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"057c8b341a51f2532ee59e7fe45286bc3367ec72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 03:57:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5681d4ea112944ee9143a116422ff501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f88268daa84deb81951aa057fe61f0 : SUCCESS in 7m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8fa42113991c41f08c2f0814c3c51a81 : SUCCESS in 10m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/710931e0a03e4fa28ad1fe197d89adbb : SUCCESS in 3m 55s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a0c5d2346c8c4cdea74f0c157dd56cf1 : SUCCESS in 5m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1ad812a97a264387964b6a632706bbe3 : SUCCESS in 1h 08m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3fba43a5c6a1408abca13ce64f6de6df : SUCCESS in 1h 14m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/391f77e7ca78421baf7586d7264b555f : FAILURE in 1h 06m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/47191e01b7fe4860af57934dc3d1f90c : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2},{"id":"475cf39bdc345168a7ccd005358b835d27692314","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 04:20:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44f9df3ab840257df690876389473d6edea5e00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 05:28:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5fffb6fa949446c9e4cadb515a9151d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e52c4e6023428382cc31619c19e0e0 : SUCCESS in 6m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bfacbaf8a2e34dfebc9f951f7d45bb5d : SUCCESS in 4m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c2beb45dff514477b74cf11eb37d28e7 : SUCCESS in 4m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d6fbe16445b243c4ab0f35b502e5ac91 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9c9d68475c6f4a1f967575fcb1988366 : SUCCESS in 1h 05m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1580b0f79dcf4c938eeb02b7eff95b4a : SUCCESS in 44m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/27474cf692b5425f826b9275ce35b88d : SUCCESS in 52m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/415499f06ff14109ba587ee7fcaeb9a7 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"dd0736502d26e0e4b285902fc863e172763a8481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 05:28:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d61ad4192043185ff4c15e53f4f99245a0657304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 07:00:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c887ff8d54b94ac69f78a559755571cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8c587beced43548afd875f9173a793 : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f0566563356d450a9b22c66a1a434f73 : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/78dc2333f7474482a4f8fe34c2b4c09b : SUCCESS in 4m 55s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/34e9a830ead74ec19400649c94679bb7 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af2ffcf7a283469982537b9fe72f5e58 : SUCCESS in 1h 25m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b8a83aa4db0f4067a23329dd1186f74b : SUCCESS in 1h 00m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f57223d428554968ba75392bf688d9a4 : SUCCESS in 54m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cd3169c1599d40c1a98b88f562e5c1fe : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":2},{"id":"36f21c722edfaace1a3cfa288782fdbdf481e243","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 07:00:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dd45e5c48da8edb7e3ab556111693000cc54966b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 07:02:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02edb0eca3d242a1adcffa7e01c4815b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100761fd6a5d4fa9a29e60ca1d95ae19 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2c5122fd91fd56b8b595df5d2220655066a70b9","revisions":{"b4d312a324d110f7de0183bf34609620210b9a40":{"kind":"REWORK","_number":1,"created":"2026-03-17 11:03:38.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/980897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/980897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/1"}}},"commit":{"parents":[{"commit":"e750db0075946469efbc076bab4953372be4ab85","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e750db0075946469efbc076bab4953372be4ab85"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:37.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:37.000000000","tz":0},"subject":"Update .gitreview for stable/2026.1","message":"Update .gitreview for stable/2026.1\n\nChange-Id: I26f6db2c2796e9b2e93e87cd93d3289015f9fac1\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4d312a324d110f7de0183bf34609620210b9a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4d312a324d110f7de0183bf34609620210b9a40"}]},"branch":"refs/heads/stable/2026.1"},"a2c5122fd91fd56b8b595df5d2220655066a70b9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-30 01:03:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/97/980897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/980897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/980897/2"}}},"commit":{"parents":[{"commit":"e660d950a6c4f91c0f6cf692b536a8eb26a1944f","subject":"fix: add IPA_DOWNLOAD_BRANCH for Ironic e2e tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e660d950a6c4f91c0f6cf692b536a8eb26a1944f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:37.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-30 01:03:11.000000000","tz":0},"subject":"Update .gitreview for stable/2026.1","message":"Update .gitreview for stable/2026.1\n\nChange-Id: I26f6db2c2796e9b2e93e87cd93d3289015f9fac1\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2c5122fd91fd56b8b595df5d2220655066a70b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2c5122fd91fd56b8b595df5d2220655066a70b9"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
