)]}'
{"id":"openstack%2Fskyline-console~980898","triplet_id":"openstack%2Fskyline-console~stable%2F2026.1~Iee05025f498541ffa9b44c0dcf8175c1ce3dc54b","project":"openstack/skyline-console","branch":"stable/2026.1","topic":"create-2026.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-03-30 08:17:22.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Iee05025f498541ffa9b44c0dcf8175c1ce3dc54b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","status":"MERGED","created":"2026-03-17 11:03:39.000000000","updated":"2026-04-03 23:55:01.000000000","submitted":"2026-04-03 23:54:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980898-create-2026.1","meta_rev_id":"cb4da726853254a744ccc0c435df5865c881eeb1","_number":980898,"virtual_id_number":980898,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-03 23:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-03-30 08:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-03 21:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-03-30 08:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 11:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-21 19:33:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-21 19:33:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"a0921c6693498436773bf2f42987fc420f59d195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-17 11:03:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f68a1b5b97c7439b8bc4aa62fc933687a62759a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:27:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0ac8a341044267a2ba7af861461a70\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e286266465674ac1b66f82944c833036 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96d68741ac24a9f8c958177b56d7e6e : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3500f391762243b8a6e72da03d696e58 : SUCCESS in 4m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ef07b47d30c44dc7b78356e8ead8c3ab : SUCCESS in 7m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8f8823bcdd3b4e47be1d81daa82950e0 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"448eda0eee80eeb9878d540e82dddf655df3df90","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-21 19:33:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dceda9bb5e63824da9059ce478afbc1b50ded576","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-23 06:36:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed5d848291bf0668b82f7ce09538831cc350a18e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 06:37:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"898dff5e9f446e4452e71d8679153ae998dda395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:16:23.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/skyline-console/+/980897 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"6a31d2b1d49fb079ea7a2856cc4cb62cef8731b0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-29 23:14:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68a474c52b1b0c3ea921c708a24beca30cc41acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 00:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/799fb5a425e74a99a258c25879febb1d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ffecf17e60f42a09a6dd347262e3e99 : SUCCESS in 1m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66baa0e48424f4092bd0b206232ea10 : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3a2af3348b81461db5047d2ee659f98d : SUCCESS in 4m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b79af5887e014357867bd0111e5c99a7 : SUCCESS in 3m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e894dcc47c7c4db1b27f35b636182a49 : SUCCESS in 8m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/42f0d1c43e1540c2a50c2123c01a2add : SUCCESS in 50m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4f6fd463a408496dad02106c19a58607 : SUCCESS in 1h 15m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4b256313b27743fa8c93517596c9a4eb : SUCCESS in 52m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9f3f50fbb4e442719eb89dc0b8bd9d91 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dfe25f8a0ec393253eb1574fdfcd05bd7f7e39de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 01:08:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"132480a4214cb9ae9525f8634e9261754c104058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 02:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7514097218164d0b82e25190eb06ebb5\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6f2a614ad3342c8948e6e253fb9d2a4 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9448a88798984bb3afaf56350ca4f608 : SUCCESS in 4m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b79f2f94775d491d883fff399f326591 : SUCCESS in 4m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f1ad1513904f409abbf056b701c5df14 : SUCCESS in 7m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a9024eb031047b8859a629a27ff2dec : SUCCESS in 3m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/33f974346d484c0389ff259984dd4ad2 : SUCCESS in 1h 21m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6fb87d4345764d01beab513af12c25ab : SUCCESS in 1h 03m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7025b5bd313643b0ad7ea8f667ca78a1 : SUCCESS in 38m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/607bfe1f3c1a4bd3a632bafbf8393f49 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f0fee541c4b73ff3e35b8590809f4df1577619d6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 08:17:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13b4b5028d3e6ea2c22fe4190508dabdee4f20c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 08:17:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d342ec0ba9d4d3d2dcef2f1c4815fb15e8cdaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 10:47:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d241e36ec34b169cbf8ba9e5a85f5b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/401a47a68de941e5805bedbcb9ebaa7f : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a8779dd51f04d36b4662f5188dd3017 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e6c08afe3bd04168b768163b971c088b : SUCCESS in 12m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cc5d31c854ab4346ab6e68006494fcb1 : SUCCESS in 8m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1cab4b6b6e724d7ab408cc19137c7a35 : SUCCESS in 7m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e3c4121ece794a0d9ab36ffc0093ea32 : SUCCESS in 1h 24m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e0c6e62ce99844628467c8770533c9c6 : SUCCESS in 2h 24m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c4bfbbd3393c4317b57a550ca6c79260 : FAILURE in 1h 03m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2162538806954a3b90cc66f64affd673 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":2},{"id":"41cfe1ad123d12ea222a59f758249f15e6b4facd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-30 15:55:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fe3df9c8ed975a108ae6724a184cb2154ccc5be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:26:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fb8b23c3a1f4f5e8742c411ae5fee95\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50c2c10d38e04e61a30c7f38ef6fb889 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daa9c9447354e8bb16156e0ab0cdd16 : SUCCESS in 6m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/37498628426e46d4967f7f38a6cc7859 : SUCCESS in 13m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c297b6b330864d27afdce5abd90a5956 : SUCCESS in 3m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/970ef3bfebcc4537a5a65eb2e8c6493c : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2038331217634953b8f07ce684243f40 : SUCCESS in 1h 17m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3eae2201820f4366bdf238d26cd54c55 : SUCCESS in 1h 24m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bdfc3fd026584de0b13ff36dd406ef5d : SUCCESS in 50m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0aa9abf5d07e480eadac920cc5f4a6cc : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2b6fd2c40b67fdfecced99a1934941df2d8f5113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:26:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94d4802f14b41f0341f608cb78536b24f2c0873d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:45:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b37a58fdb68540359fbfdd9c136533fa\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a44a020bc194716a6f94067ab837ba9 : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692655655a8849809cb032c4cbb9c435 : SUCCESS in 3m 20s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ba493c2c8e68424b804c061d51958ca6 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0be0be37a900498c96a6964258e62b41 : SUCCESS in 3m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ef65fbbaf28643b59a86cca112dfb704 : SUCCESS in 6m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2ee3e5f3092640928c3ec233af98c673 : SUCCESS in 49m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/47a082d9695e4061bf3ad4e6f20853b0 : SUCCESS in 1h 17m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7cb21a0a5da5466bb925d30e0bba3e56 : FAILURE in 1h 12m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fb38ff343afa4febb6050c75d80d1b4c : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":2},{"id":"116a81e9f0f4343d815a7cfb20f7cf44d967c354","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-31 07:54:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8c965320e1ca4a46d0f384f1a30c11eb4352c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:35:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c4c1469a594fe583a65ada833b6db3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff37ccc42aa84112bec9c8abc709cd81 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7dab596b9d40beae4aa4431c736337 : SUCCESS in 6m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5171154d57cd44d0830532dacab129ee : SUCCESS in 5m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fde2b501298747738070bd6842bbe4de : SUCCESS in 8m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9087c307852848398f7fe551f55f0624 : SUCCESS in 5m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0d96789bab074c969c233e2d83cd5d83 : SUCCESS in 1h 08m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0dae44751e694a63b7c0f6f646dd58b6 : SUCCESS in 2h 39m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0bd34b48de0a40beb711dea09c6199b5 : SUCCESS in 42m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fa3b1e5a641a4a70a1ab957613e0b685 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1a7ad2c90907dfd4421b79d13f3c9a27a1c99a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:36:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e6c804f6e49a85a30b7af68daf0df289d2bdebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:31:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff3604d24844c6889d511445357f1fd\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3005a512064977923fdef2e3478d22 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b36375e390440bbfedbc18160b31eb : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dde4a8ec8f59405b9df4b0d9602c24d0 : SUCCESS in 13m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e30511ac4c3146ecbc66edae9c47542d : SUCCESS in 7m 59s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/42cb2acce95c4af3878bbd53312837e6 : SUCCESS in 7m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ef786f4b80a84d81a2c7c96fa0ae5e0d : SUCCESS in 58m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4ed8c37d1dd44f0f816d88a9f9000beb : SUCCESS in 1h 54m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9c24629f81694797b2df719ba13234fb : FAILURE in 42m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e658ffc8c2494627adee0e59e9e13328 : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":2},{"id":"518db406137edf1017cac263ffb1397edcbd44d0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-31 14:21:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c535f26ca6e6b7e3d96638576fb84265838779b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:29:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04badc9cf8b74d7bab2e0741fd6b2518\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3354ff3d6d1143e8af8615edba56b2ae : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09866681b10e4acf8a0e60ed7bfdff8e : SUCCESS in 5m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4d102804d0584bd386b94a9f01106bb0 : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7dcf1060c0094ab1b48666408211c7ef : SUCCESS in 5m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d482175cb12747f6a7d24b688b0a806e : SUCCESS in 5m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/170f23461ef54f96b1584f5d4fbd2ad8 : SUCCESS in 1h 23m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b28b62f3f07140bfbe9e0c6d35cb02c5 : TIMED_OUT in 3h 05m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e03599a453794ff0a0d3a0d41cc4e1c9 : SUCCESS in 55m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9f18aac12e4f4a6da4bc4d95993b95e3 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2ba7e9902e3b30cf13fc9d0c634bfed18188c5d5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 05:51:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ca79558a6aa5a7cf1b9f68fcb2fc1b83b49db79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 07:17:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9fbabbce2544b48552876a1d7cad2c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b280e6bb07e84165872037ad9f859bb2 : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e442f646f2947ea9d2b77b77dfc510f : SUCCESS in 5m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/72f0daf702524e1c9a4500e8c8ff7724 : SUCCESS in 4m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2eb2356f5f4a4d15aaf9bb4817170724 : SUCCESS in 8m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b794314561724cf38d46d750ea631319 : SUCCESS in 6m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/78b76b24b38c4fc7979666ac6a64fb8e : SUCCESS in 1h 21m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4cb718e0b7eb4ab28e939f13d9b5f219 : SUCCESS in 30m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c8b5ff6368bb4d4ba1717f9b83519104 : SUCCESS in 52m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fc947973c2b64f738a40a6aaf9a98b70 : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f2c49b29e352db3cf84b081ef1c81d93a51458e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 07:18:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b60775198112c1fd7f20e9b36e60d9ed5139517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 08:28:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae363ef8f5cc4500a4e9d15f2b93a936\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6296b6b6b69a46678e7ce02084594c82 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2aafdd9eaae47218e234b0261c69bb6 : SUCCESS in 5m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a429bd46bd174ac7987737251575b8f5 : SUCCESS in 6m 39s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/da663c10f75843788d27feeb842d9b3a : SUCCESS in 7m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bbc60306d8c945f5a2a31767efa31a10 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bf594f6844674454ba17dc70d1e21298 : SUCCESS in 1h 08m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4d23014eb0ac41f8b5d06e9fa2727d97 : SUCCESS in 38m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/47e118646c9b41b2ab481da9bff20993 : FAILURE in 42m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ee9698939d1e4253b253171a98329f22 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":2},{"id":"1998037db02af03ba4833c72da736e076b217771","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 16:46:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9ae7258baa9c40c1b7404304c8411017b11927f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:23:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5879abb1d0364ef390282e6915ec102f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cda12dda09b4f85a2bfa1c8db5abe89 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13cc124d66a407ba739d65161a756b3 : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5cd5ab1030c84e5f9a88b1896723022a : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/35731dfcdbf84a858e74f453817269cc : SUCCESS in 4m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0b2c1c5c3c734c579e16aca9bfb745a5 : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d5833ff5157c48178bae8d06b0e7fd1e : SUCCESS in 1h 30m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/131d17cd767f497a9683d787474778d0 : SUCCESS in 33m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/dd701828cf28473284308bb6c35689c5 : SUCCESS in 54m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6c69ab07c459460b9682123163f4ea01 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":2},{"id":"28bcaeb7da94345cd6c540486f67fc90048cf623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:23:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"018957ce252a64aeb5ba36afb5791842b5dca898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 19:29:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f4ec3db48a34d99abeec7a63eab121e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bf6d2babef04820b19ebca9424583db : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f2fb37a5574da8a12c771af2373552 : SUCCESS in 3m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cbd2f9091d234c518170dab9ba131d8a : SUCCESS in 12m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bd88bd0a903c4ed2ae88e6bc4335f0ca : SUCCESS in 7m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c7f06c35f2e446bf935a8fdc0aba69b2 : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/79114a8a59b24c6b9c59b6929885b2c6 : FAILURE in 54m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5276356362c14857977d5d0bb1319125 : SUCCESS in 1h 01m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/671dacdccac24d75af8f1d70cbb50b3a : SUCCESS in 53m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e8725311b35f46378270be834ca1aa18 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cd08fb96b4d287e88841d5f98388e3f9bdd8a9f5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-02 06:30:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a996793e3c1474aba1d8922e9278fdcbd9b2871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 07:39:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f62611f08c1b4a50b294403e9142746a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02ac01a9a2314d1890555692e9703660 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565320dfe1f341caba52b254ae7f45b2 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4290cea1b6b44953a418f7e2fb49be37 : SUCCESS in 11m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bf8c3613641d4e02ba3d33e42aabfecb : SUCCESS in 4m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/035541d7937146af95097d3cc0bdbc30 : SUCCESS in 7m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ec75de858acc4f049b2f23fac3a8c975 : FAILURE in 55m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/42cc51e27751418c8be80511faf64c41 : SUCCESS in 1h 04m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ecbd3a28bd6e4f2e8adeb7ae717a2664 : SUCCESS in 56m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/92ab3071a65f4c4c9bfc57d2248a6b81 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aff1e635d6a0651f8175355b572f33adc1099899","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-03 21:18:58.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce9868099959577a4afdad9fdbc55111855dc189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 22:36:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0096f4d877b482bae1fdcc3df8fc642\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85f64a4ebad645d9a506d1c5cc47eec2 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c7015cb3aa42d49e654ad7ab7d472e : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ebecb75c84ba464b9aae1897fb402543 : SUCCESS in 11m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b54644667f704998b5751d1be5c1f585 : SUCCESS in 7m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/17e6e50c85c6444084afe7b05db08043 : SUCCESS in 8m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/074b43abd2814eed8de6e13a2c91baa6 : SUCCESS in 1h 11m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b51a5c0f757a42ec9a59a0125ab23c77 : SUCCESS in 1h 00m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f766d5e024ef4e3ebd4c992b9279faea : SUCCESS in 59m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b426cf243c0b4b73a2f16f4b5975c4f7 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a67667d69020664450a73fc69ada817f2e1e9435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 22:37:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77895878a881d959bdc57271b7f66e6f885dae3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 23:54:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4160bc992c5949a0a9aedddb03b1e95d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/922359d9f98b4160974e8c292db3426c : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf70610463b4ed78aef00b7c6fc26d8 : SUCCESS in 7m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74d3aba3c7d547eb9561834468dac9c2 : SUCCESS in 7m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f9385b099e5b4e11926c8508f1f6c9bb : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/306f632089a1495fb914276d662032ea : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a4d80317260540cfbae467c978787ace : SUCCESS in 1h 11m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/144801b59db7459ca3cd5931711cef9c : SUCCESS in 1h 01m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/17ae203a0ece4fcd8a8481f7eed5e734 : SUCCESS in 54m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/985c25bc3f284e5e8caeeb15fdd699d0 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a4297d8c8ab54da9cebb6741096ba806fc95e9c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 23:54:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cb4da726853254a744ccc0c435df5865c881eeb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 23:55:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060d6e6d9cb14ff0b8c60210ad33ad5f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51f62f1a505d4b3dba15a8a4f2f9877c : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2e6924080e4e6bb769de6b0dda7a2d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97112520c9361ed9be83b845d6737c4eacaaa0a2","revisions":{"9f65c9f40d8b6eec5cdbfc033ca4398088e15a79":{"kind":"REWORK","_number":1,"created":"2026-03-17 11:03:39.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/98/980898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/98/980898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/1"}}},"commit":{"parents":[{"commit":"b4d312a324d110f7de0183bf34609620210b9a40","subject":"Update .gitreview for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4d312a324d110f7de0183bf34609620210b9a40"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:38.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:38.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Iee05025f498541ffa9b44c0dcf8175c1ce3dc54b\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9f65c9f40d8b6eec5cdbfc033ca4398088e15a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9f65c9f40d8b6eec5cdbfc033ca4398088e15a79"}]},"branch":"refs/heads/stable/2026.1"},"97112520c9361ed9be83b845d6737c4eacaaa0a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-30 01:08:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/98/980898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/98/980898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/98/980898/2"}}},"commit":{"parents":[{"commit":"a2c5122fd91fd56b8b595df5d2220655066a70b9","subject":"Update .gitreview for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2c5122fd91fd56b8b595df5d2220655066a70b9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-17 11:03:38.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-30 01:08:25.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Iee05025f498541ffa9b44c0dcf8175c1ce3dc54b\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/97112520c9361ed9be83b845d6737c4eacaaa0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/97112520c9361ed9be83b845d6737c4eacaaa0a2"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
