)]}'
{"id":"openstack%2Fskyline-console~982350","triplet_id":"openstack%2Fskyline-console~stable%2F2025.2~I5704697f2c8bd0321014844cadea202911ec291c","project":"openstack/skyline-console","branch":"stable/2025.2","topic":"2025.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-03-27 06:29:18.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36902":{"account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"last_update":"2026-03-27 10:00:52.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-40212"],"change_id":"I5704697f2c8bd0321014844cadea202911ec291c","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","status":"MERGED","created":"2026-03-27 03:28:43.000000000","updated":"2026-04-10 17:07:42.000000000","submitted":"2026-03-27 10:00:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982350-2025.2","meta_rev_id":"bfba4286e7efa1e79b2f6b22b6edde3b6991c248","_number":982350,"virtual_id_number":982350,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-27 10:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-03-27 08:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-03-27 08:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 03:28:43.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"state":"REVIEWER"},{"updated":"2026-03-27 05:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cab478bcda32cddb6a724436aa20a72ac720f969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 03:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbdef096e924ac8b7da4785be02bba3ba04ba19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 05:23:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7e27bfb5244b79bee7bd7c35d93c42\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b46c9ddf1b4fd8a5090abed77a9256 : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748ab78ea57445cdbd05b36bebc1d450 : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/adac3ca59f2248e49545805e58067c0c : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f3afbd18c29448c9b082660c57963192 : SUCCESS in 3m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5b12292a1c2040988e9c9d4fbcbeac44 : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/403c12641bcd41979b1f1e8ce9729753 : FAILURE in 54m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/97ae1ef31ae2498989c784ee7ab678c4 : SUCCESS in 1h 53m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d4484867196b443582dc3b7410235e8c : FAILURE in 42m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f879cc952c014f33909256ec7517ffd2 : SUCCESS in 30m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cb3315639d25d2ec763e28bbd4ad1f86d21d85c5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 06:29:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fd8d04e788966f68e0e4831ed2ef0912fa7de6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 07:57:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a1c7c8fa45347ebb4bb3713316552b0\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5ac5cda219947d09e669015ab2fbfe7 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2969fca7073d4ba0a122c6cf74f2d01f : SUCCESS in 3m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b6e656d010164937802ad47008a1ea08 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7f771c81ac66454b8664d943316c442e : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a55ab140b2df4dcf9038819dc048e9dc : SUCCESS in 4m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/45259e5d96574cd6b74ade507b29e854 : SUCCESS in 1h 22m 15s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c3b40224522f499f892a7a87f9ac93eb : SUCCESS in 32m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c691586779eb468fb6ed1865030cabfe : SUCCESS in 32m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5ce96379676b4c4b98f9173476b246f9 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":1},{"id":"56a6828a148336d7c1f8a1a885651684d62f03cd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 08:49:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23133869ef8c4926d745e91f4b52c8d39f9dfab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 08:49:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22f51253d24cc8436faa62c1396e731eb49bc20a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:00:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863e33a62b2b472787d7de73e8cd03fd\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096c0029b6834647b52c338050afc63b : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94843f738d3491cb828aba4acfcf233 : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a7d37d3e41154132af4614e0097ff938 : SUCCESS in 4m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/71e560fc2d844b1ca97c4540fb5f79d4 : SUCCESS in 5m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f44934de38744e7cb8fa1af5f6e3fb08 : SUCCESS in 8m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a3ff759cbb334fcbb7001e6d556931c3 : SUCCESS in 1h 09m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3fe008622e4d4c818f46d50b345bff0e : SUCCESS in 31m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/440b75c6353b4bedaa89aacac2c99390 : SUCCESS in 37m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/63bb293ca2484eb5b4f3a997ee12615b : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"da144854440459427c0c87f76989d57ec06e1ac3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:00:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a92299674744722a9214a27f5822320f29b05cf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:01:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8925aaa694744cd8199c734702a3846\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/098a632591de4e19bb1027dda2059db2 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337a6c296db44bf7856eddd6786aae80 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"bfba4286e7efa1e79b2f6b22b6edde3b6991c248","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-10 17:07:42.000000000","message":"Hashtag added: CVE-2026-40212","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c8397afc56a2bd77758ddd43142e08c0238eb50","revisions":{"8c8397afc56a2bd77758ddd43142e08c0238eb50":{"kind":"REWORK","_number":1,"created":"2026-03-27 03:28:43.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/50/982350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/50/982350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/982350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/982350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/982350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/50/982350/1"}}},"commit":{"parents":[{"commit":"cd4d79b83e460863bc73b9687a1243b1c587e3ec","subject":"Fix IPv6 subnet fields visibility in network form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd4d79b83e460863bc73b9687a1243b1c587e3ec"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 03:28:21.000000000","tz":480},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8c8397afc56a2bd77758ddd43142e08c0238eb50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8c8397afc56a2bd77758ddd43142e08c0238eb50"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
