)]}'
{"id":"openstack%2Fskyline-console~982355","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~I5704697f2c8bd0321014844cadea202911ec291c","project":"openstack/skyline-console","branch":"stable/2025.1","topic":"2025.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-05 06:43:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36902":{"account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"last_update":"2026-04-05 07:59:35.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-40212"],"change_id":"I5704697f2c8bd0321014844cadea202911ec291c","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","status":"MERGED","created":"2026-03-27 05:35:06.000000000","updated":"2026-04-10 17:07:51.000000000","submitted":"2026-04-05 07:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982355-2025.1","meta_rev_id":"0793d471aec78414bb05abe6c2a9f89968b713e5","_number":982355,"virtual_id_number":982355,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-05 07:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-05 06:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-05 06:43:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 05:35:06.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"state":"REVIEWER"},{"updated":"2026-03-27 07:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56a177ad8ea5be6219d672f93d27a8ad2386266f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 05:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b13d25f6074b7e5b572fb56c7cde5a5891f6288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 07:37:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03176e65c2404a88abaaa016eca4ca4b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db48ad28948546ac8147c09cc106a783 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cb7a039eba438daa7443375e867a96 : FAILURE in 4m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/823ede9a3a3e4861bd06279d377fdef1 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0af5f8497c114542a510ada13c4fb5e3 : SUCCESS in 3m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d99656640260495d8f055665518c3ef6 : SUCCESS in 3m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/63dd3f755a9a477fb416499699e9b1e0 : FAILURE in 2h 00m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bbb0997aaa404ddfb2f27405811aa21a : FAILURE in 59m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e4f22b0296434211b75ab965aac84be4 : FAILURE in 54m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/59fee129375e4a5d94663f700bb45637 : FAILURE in 1h 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e36218d9ffcb129795510354a9bc46af23ed690a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-28 16:22:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1accbfd661a1af3a052811c89859289171224984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 18:37:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b050955e9794c24be7d72787e2010dd\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/921dc43048904087a3cff4d6060c4dc9 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c21c06dd5145518ea5ea7671edac36 : FAILURE in 5m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a0a4f5bfa2d1432587556c645cb57828 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/08316195f05542e699e93110725f9b0e : SUCCESS in 7m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9fd980569ed647f6b0d62bae48512b0f : SUCCESS in 4m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6191d12ce4bc4ff2bfb3d35e3cf59b17 : FAILURE in 2h 08m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ca536f1fc3d541e5b18c816d53c251fe : FAILURE in 1h 41m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cd07bbf2b3a94ef6a523f21ff302c66a : FAILURE in 59m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b4d9794535ea45a5aa5ef3c472aeef51 : FAILURE in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6548e7085def8c32f809a9a235bb0aedefe2251f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 04:01:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68b59b1f0323e9613342eedc4f8ca297ff26d4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 05:26:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26faafa7289a4fc194f98f27a3d483c4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c93b119c2e3f44e697f1bebaff917cc7 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb9264b56604a90aacf6c6763e2681a : FAILURE in 4m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5134cc2efeab4ecc8780a83f68a21a7a : SUCCESS in 11m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/24b68c137cb44d17ab56620f25a6eb9f : SUCCESS in 7m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2c744a5d49634897a7f99f834b4ff3f1 : SUCCESS in 6m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b565d80ab3534552a36ad6b5f5de3f80 : FAILURE in 52m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e3bf8339bf0b4ec399afdc77a9231dc9 : FAILURE in 1h 24m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3893310b875a4630893fb35d369b801f : FAILURE in 43m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0e972df0db1f4ad18b3a3d4b0cd02299 : FAILURE in 1h 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5bbc19149d3bf1728c679d1785d6808a30169c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-02 06:57:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8ac919584d4dac86d2627316bda7e5e83c50fe91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 08:34:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77727ad270304d4090c1f7e47eeb7500\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe3268013b54b3481ea0ef762a16a4e : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eca7b99704485baada0779160942b7 : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c6fa1dc4ade04b9d8e1cbdbccbcfc10c : SUCCESS in 4m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a9298e7343b142b288c8aac392545c12 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/21a5c3ab88fa434eb500a49d71b3c733 : SUCCESS in 4m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9619b7dd526b4ae2bc12cabf36914cfe : SUCCESS in 1h 32m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/802c3c64d23d464d9a93ebbb50f9f184 : FAILURE in 34m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a7e0ffaf9df74b8fb7a2c7af8f12b116 : SUCCESS in 53m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1e02b6a03f1b454a84bb2437fbb91a0b : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":2},{"id":"936671aa7420bc21a3a80cf15ad35f5609493614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-04 16:23:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"346928567b709b882e337c3198afcb6e36f7c7a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 17:38:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3398d511fb9473ea0e27f976ab8f9ce\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc9f26e21204e10bf61a068dc60f7bb : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a7de4786514e8b836b344d1a613ab8 : SUCCESS in 3m 43s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c1c800082a184dfa8bc461e5fdcf81bb : SUCCESS in 4m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1c39c8485adf4fb5b5f2e4ae7ce5ee1d : SUCCESS in 7m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1e5eddf92539483c90afbbffb2c132aa : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b3635a281cb6438f8806a0727c087834 : SUCCESS in 1h 10m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4480b4d7a987450e8d7f9ddef2f02c53 : SUCCESS in 45m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eb89ba98b0c14eb998bd2a8c1b84e7d3 : SUCCESS in 40m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/60ffef708d3749e597d767fca761c0a2 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":3},{"id":"8c70bb191d662129acc9944c085edca984abe26e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-05 06:43:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"20b84914a45e5493dc28384e0dae09432ff1a0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 06:44:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8c4ae61072adf462090ff5ea935771117a1836c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 07:59:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f342f6620f74bf6aeba43b3d12c57b8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1def729d504480694c3fb88468ec60f : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dfaec0c8ec402b8dae0a948ca503dd : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/93f2c4f837544935b0184961ebbb3442 : SUCCESS in 7m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/defc41e03fc0431fa28f3c8d39821f83 : SUCCESS in 8m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f7bc6359d1e3460995e519f39d2eb59f : SUCCESS in 4m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4f2fc659e84140cc8c3280bcb86472c0 : SUCCESS in 1h 10m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0c9395b0acef4e5ab8e1377a41da03c6 : SUCCESS in 1h 01m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3f53a3d4f1e54ad5a10040728c6a511a : SUCCESS in 40m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/398984d54fb44a8a9880f1dcf4499acb : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d04cfc19122494785ca658fe8d3189a94d2059c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 07:59:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2d424df788a36d7e30dc5cacd9b729e185baa10f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 08:00:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239b791fc3f04e9b8c25ac5d858601d0\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4a66482fa974a64bc428f28d02d56c5 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f4f512470f4d7ea65b4086ad264531 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"0793d471aec78414bb05abe6c2a9f89968b713e5","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-10 17:07:51.000000000","message":"Hashtag added: CVE-2026-40212","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ca1bcec9d3b318e2b710d05e40d0d1edc82e868","revisions":{"fff1a3eaf50af07f029b6c93a0eb4b389a8f8e13":{"kind":"REWORK","_number":1,"created":"2026-03-27 05:35:06.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/55/982355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/982355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/1"}}},"commit":{"parents":[{"commit":"46fc2803ed3d895421adf45c1b719bad0dd930c7","subject":"Fix IPv6 subnet fields visibility in network form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46fc2803ed3d895421adf45c1b719bad0dd930c7"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 05:34:28.000000000","tz":480},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fff1a3eaf50af07f029b6c93a0eb4b389a8f8e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fff1a3eaf50af07f029b6c93a0eb4b389a8f8e13"}]},"branch":"refs/heads/stable/2025.1"},"1a32e6a92a1fe91413ce7169b5d4593cd74a1fdd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-02 06:57:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/55/982355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/982355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/2"}}},"commit":{"parents":[{"commit":"82bc481f01330d4f50c2e5a60e8a3d86cd5adda5","subject":"fix: add setuptools for pkg_resources in doc build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82bc481f01330d4f50c2e5a60e8a3d86cd5adda5"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-02 06:57:20.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1a32e6a92a1fe91413ce7169b5d4593cd74a1fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1a32e6a92a1fe91413ce7169b5d4593cd74a1fdd"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"1ca1bcec9d3b318e2b710d05e40d0d1edc82e868":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-04 16:23:12.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/55/982355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/982355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/982355/3"}}},"commit":{"parents":[{"commit":"6c1e1d979912d2b3f1b8ad0b48264277132bd14a","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c1e1d979912d2b3f1b8ad0b48264277132bd14a"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-04 16:23:12.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ca1bcec9d3b318e2b710d05e40d0d1edc82e868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ca1bcec9d3b318e2b710d05e40d0d1edc82e868"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
