)]}'
{"id":"openstack%2Fskyline-console~982356","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~I5704697f2c8bd0321014844cadea202911ec291c","project":"openstack/skyline-console","branch":"stable/2024.2","topic":"2024.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-07 05:24:30.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36902":{"account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"last_update":"2026-04-07 05:35:31.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-40212"],"change_id":"I5704697f2c8bd0321014844cadea202911ec291c","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","status":"MERGED","created":"2026-03-27 05:37:04.000000000","updated":"2026-04-10 17:07:42.000000000","submitted":"2026-04-07 05:35:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982356-2024.2","meta_rev_id":"5892322f31496d04a21d4568b307fa8d10d23b88","_number":982356,"virtual_id_number":982356,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 05:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-07 05:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-07 05:24:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 05:37:04.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"state":"REVIEWER"},{"updated":"2026-03-27 06:25:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3adf0dbea07f4ce19b577fbf0931a6f55e47f69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 05:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a28828cbf96f2d12fd0e3acfcea89985079d4cad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 06:25:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b260f30086b47eb94035e7a616a6960\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af9900fab8184a9ab0ed335b9c335a80 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af68d82cf0844268cff8075021dc994 : FAILURE in 3m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cd5693a01b8149d8aa827fd3f6868a5c : SUCCESS in 10m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6784d0d0b639462fa09887966cf8c83b : SUCCESS in 6m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2a1d469eb55b4eaea29d1e7175a3a297 : SUCCESS in 4m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/aa086e634cd2431d842750a8e382445f : RETRY_LIMIT in 13m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8a46ee33aa6a4b8d8118212903ad5adb : RETRY_LIMIT in 11m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bb37a456b71343f4a8a15a7e49d6cadd : RETRY_LIMIT in 6m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d9b4a831fe0d4b8cb0e16984e01c5134 : RETRY_LIMIT in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"55fc7d3d6455ec7fac61c172358d727c5e3f1903","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 06:58:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45dcafb3217ac5a9a1a995ac1c66b3cb1783991f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 07:45:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9c9482ae559493c8a1c0449975270c8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae5fbd3588f749fa91777a055855a5a0 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8570d3b38454473381a25b2d86a27c81 : FAILURE in 2m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ad2f7624834444eb8e7d935546e5182d : SUCCESS in 4m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8fa48d50da2c42839919fdf791438141 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ef03fa2f59614c15a0b2460c0aea1268 : SUCCESS in 4m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a6e205a6636b4356abc009bf084208f2 : RETRY_LIMIT in 10m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/51a09dbd156a43d999f14bf5d7f82a6d : RETRY_LIMIT in 9m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/266aa3339bec4af2967943e44d9bfe2a : RETRY_LIMIT in 4m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/424bdae6b9f5443795a7eed1e0762e64 : RETRY_LIMIT in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"67db0998ca23baf3e624be4015474990e219f862","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-28 16:22:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7320788c27f22b3be87858ca34021b5c979960c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 17:10:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0f8a65a29aa4e4694ce31641b98acc5\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c7cd4987754864a148c25c4bd4dc7f : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e6ca14ec6b4d17b5b51af5ad17766b : FAILURE in 3m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6b8ee379cc4d4073a19a6771f917b715 : SUCCESS in 4m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/00b389217036494696138c4db9e71865 : SUCCESS in 4m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/70104af60e294313a943e81f490c20bf : SUCCESS in 7m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/74b618a99c864cc5895dd86b4ed9011b : RETRY_LIMIT in 12m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/233eb3a53ddb4c07a047bfc87db64d95 : RETRY_LIMIT in 11m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eb633e0fd36143969127cb1ef5ba4b92 : RETRY_LIMIT in 3m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8b873b1e12024da786c8cf43ea7f838a : RETRY_LIMIT in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"48964b6616f6b96bda605745badd7182ae3b2840","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 03:55:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c5575b6e28d05bed41e758c87579a306e2e7ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 04:27:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0dd03ec68e442759feaa450ece87f15\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d746346a264342bb88e8aec507259595 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7a7b2c917044bb89d75bf442435ce4 : FAILURE in 2m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/53dabd23e9754ca7b580be8b07eeed78 : SUCCESS in 6m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/87ec373a318b43b8b3399bf35786be76 : SUCCESS in 7m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/beaf317a1a0e4980ad7a6027bea8a71c : SUCCESS in 4m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3b9a877cfa2547ac96644d602da80a9f : RETRY_LIMIT in 7m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8743e8302f774412b95b3a9afc45370e : RETRY_LIMIT in 4m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/40e8ca77b4464675b927a139f6c375f6 : RETRY_LIMIT in 3m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b35f336017014a748f1389ae0f021f90 : RETRY_LIMIT in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"791fce7ec874c8929213de1a36612a6ed8d78bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-07 04:56:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3d6dfee82b02373972bf32187ea759a0fd096c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:12:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78c808ecdfd64838af8e902bc1d19f68\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/488862047cf64373ad52ad382829f8a4 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d171af75a0c644a6a6a9293d0c2e7b88 : SUCCESS in 4m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b35100d607744caebcf36e33b0070dd8 : SUCCESS in 4m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7c79b36e436746c6bd472db5722223f8 : SUCCESS in 5m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d33ebc347ff847b59f0c6ee83d13c4de : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e4588459a18cec16c9dc62c0f08a990afc728768","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-07 05:24:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bc9c087306e626af96ce14d1aa9ccf20b147b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:24:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b98236d1f993418e2cebb566848b6605a539a487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:35:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d467359f0b3341498ee9379cd2337262\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16b2cb82d7014af0bf06cddaac560d60 : SUCCESS in 1m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee68acc62afa436a8fa1593fa4b5499d : SUCCESS in 2m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f58bba88335049fe935774a541ca6657 : SUCCESS in 4m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/02ca872ec480471e985813e32be41f12 : SUCCESS in 5m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9fc069a75ee04b168c7e7ad14b6d7460 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"73c97c73b66722c1cb16c9d3efdb37e1e6d9b5f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:35:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1839d79819eb404cdc78dde98dbee1615cffe460","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:36:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa51e9f5b9d747d7a2ad253aa41a3dd8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03292be39d2947b9bb2114431cd1ab6d : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a522d619bbe47fb987bb5f91f2ef72a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"5892322f31496d04a21d4568b307fa8d10d23b88","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-10 17:07:42.000000000","message":"Hashtag added: CVE-2026-40212","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26d9d2cfb86a13accaccce068086c56569e3eb1b","revisions":{"0d6d35f5d99fd19f90c2dc8132aa9b033555ec2c":{"kind":"REWORK","_number":1,"created":"2026-03-27 05:37:04.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/56/982356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/982356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/1"}}},"commit":{"parents":[{"commit":"da4d16212f1e6cbfbb2ddde68722f24383a6820c","subject":"Merge \"feat: update package for release\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/da4d16212f1e6cbfbb2ddde68722f24383a6820c"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 05:36:44.000000000","tz":480},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d6d35f5d99fd19f90c2dc8132aa9b033555ec2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d6d35f5d99fd19f90c2dc8132aa9b033555ec2c"}]},"branch":"refs/heads/stable/2024.2"},"26d9d2cfb86a13accaccce068086c56569e3eb1b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-07 04:56:58.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/56/982356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/982356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/982356/2"}}},"commit":{"parents":[{"commit":"c2532d409941df8dc9d44853d11376e072db727d","subject":"fix: add setuptools for pkg_resources in doc build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2532d409941df8dc9d44853d11376e072db727d"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-07 04:56:58.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26d9d2cfb86a13accaccce068086c56569e3eb1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26d9d2cfb86a13accaccce068086c56569e3eb1b"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
