)]}'
{"id":"openstack%2Fskyline-console~982359","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~Iea4c7fb8727aaf3cfdf713c7c0bb9aba3ab44156","project":"openstack/skyline-console","branch":"stable/2025.1","topic":"2025.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-05 08:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Iea4c7fb8727aaf3cfdf713c7c0bb9aba3ab44156","subject":"fix(i18n): update locales and fill missing translations","status":"MERGED","created":"2026-03-27 06:05:26.000000000","updated":"2026-04-05 09:07:37.000000000","submitted":"2026-04-05 09:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982359-2025.1","meta_rev_id":"c9486793f42d2681052f9d08b1c115873654da5a","_number":982359,"virtual_id_number":982359,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-05 09:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-05 08:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-05 08:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 06:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c55b2d2ee2810207d2ddda6a4ebd8d357ea5f7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-27 06:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3ed1fd211e70d4f1b95691b284baaa4df628f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 06:12:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56704e0dca014aeb8ec05d7a46dda6ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b20607fdc746588f130ee8f365544e : FAILURE in 2m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/51eee01cd0f44ae48b4f3d2fdd28e2ca : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f842ac23e3824c2f999f3c01dd2f2b39 : SUCCESS in 4m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a6b522d77d0049d68b84deee25eb42ad : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"82878928d3b22c0767317a5c89beeb53684f4526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-02 06:57:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a974486c2a0e86218d30ed51e9da9e2ef5e9ffd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 07:13:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e9baf16700b4055bd9479136956decd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5189991e140c4d268d0175785f9e75a5 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ca8621523b94464498b3fd4ce415f549 : SUCCESS in 10m 39s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2802e28b2068431b8645eb95df07015c : SUCCESS in 7m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7c3c39d050fc4a1989eacc0e22e6259a : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"94c31a8f356354123213b916e0dd19b5e5bc588c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-04 16:23:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1020a4f907842029d42777cb57dbafbafd760f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 16:40:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f785814f39674894a03016530d727e6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fd0deef4ba4d42b959345a5efa40a9 : SUCCESS in 5m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c0ff735db15242cba7261c1ec27b4788 : SUCCESS in 11m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/77a9189cc510403e8c777aa08e33ea8c : SUCCESS in 3m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ccc7c1d409b9467fb81230395b60040b : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f3eeb99893f2946bc2e3fd0c8987993accad0968","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-05 08:48:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c58427c1ef89a2e3edfa398297b865510f663ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 08:48:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ccf3de9ed22cdbe843f51181252f3a1bb519aab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:06:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573fe69256354745b003c78b2d586a16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410541397f694d559924bd0466b7daeb : SUCCESS in 5m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/632da697de9f4effb8e6f6906788b776 : SUCCESS in 11m 53s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/eb6e2f9fb49c404b87905ba8b91ed37a : SUCCESS in 6m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9d434e7172b84e178d1fdcfdac23111f : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4345a6d1caedae6f42d7e62bb0fdcb22a470ea79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c9486793f42d2681052f9d08b1c115873654da5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 09:07:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4d95c7ab3b48c19d9aed6f0309f826\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1896888d8a3403bb6a44c5e5ad038f7 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"efccfbdcc88d3c69d54db7af7a048a325193bff1","revisions":{"3ba888bccbfcb431c248503dc724d95a2286d0d3":{"kind":"REWORK","_number":1,"created":"2026-03-27 06:05:26.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/59/982359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/982359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/1"}}},"commit":{"parents":[{"commit":"fff1a3eaf50af07f029b6c93a0eb4b389a8f8e13","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fff1a3eaf50af07f029b6c93a0eb4b389a8f8e13"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 05:59:10.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 06:05:05.000000000","tz":480},"subject":"fix(i18n): update locales and fill missing translations","message":"fix(i18n): update locales and fill missing translations\n\n- Add log loading strings to en.json\n- Sync new string keys to ru.json\n- Fill missing translations for zh-hans, ko-kr, tr-tr\n- Remove deprecated \u0027Console Log\u0027 entry\n\nChange-Id: Iea4c7fb8727aaf3cfdf713c7c0bb9aba3ab44156\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ba888bccbfcb431c248503dc724d95a2286d0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ba888bccbfcb431c248503dc724d95a2286d0d3"}]},"branch":"refs/heads/stable/2025.1"},"ac7004d513ce6d87a988e30c0ab1716f79cd1e66":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-02 06:57:48.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/59/982359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/982359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/2"}}},"commit":{"parents":[{"commit":"1a32e6a92a1fe91413ce7169b5d4593cd74a1fdd","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1a32e6a92a1fe91413ce7169b5d4593cd74a1fdd"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 05:59:10.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-02 06:57:48.000000000","tz":0},"subject":"fix(i18n): update locales and fill missing translations","message":"fix(i18n): update locales and fill missing translations\n\n- Add log loading strings to en.json\n- Sync new string keys to ru.json\n- Fill missing translations for zh-hans, ko-kr, tr-tr\n- Remove deprecated \u0027Console Log\u0027 entry\n\nChange-Id: Iea4c7fb8727aaf3cfdf713c7c0bb9aba3ab44156\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac7004d513ce6d87a988e30c0ab1716f79cd1e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac7004d513ce6d87a988e30c0ab1716f79cd1e66"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"efccfbdcc88d3c69d54db7af7a048a325193bff1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-04 16:23:32.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/59/982359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/982359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/982359/3"}}},"commit":{"parents":[{"commit":"1ca1bcec9d3b318e2b710d05e40d0d1edc82e868","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ca1bcec9d3b318e2b710d05e40d0d1edc82e868"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 05:59:10.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-04 16:23:32.000000000","tz":0},"subject":"fix(i18n): update locales and fill missing translations","message":"fix(i18n): update locales and fill missing translations\n\n- Add log loading strings to en.json\n- Sync new string keys to ru.json\n- Fill missing translations for zh-hans, ko-kr, tr-tr\n- Remove deprecated \u0027Console Log\u0027 entry\n\nChange-Id: Iea4c7fb8727aaf3cfdf713c7c0bb9aba3ab44156\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/efccfbdcc88d3c69d54db7af7a048a325193bff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/efccfbdcc88d3c69d54db7af7a048a325193bff1"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
