)]}'
{"id":"openstack%2Fskyline-console~982548","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~Id7c86d1093314f6e4e8b9dd8b106df883b671ce0","project":"openstack/skyline-console","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-30 11:53:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id7c86d1093314f6e4e8b9dd8b106df883b671ce0","subject":"fix: invalid vars syntax in playbooks","status":"ABANDONED","created":"2026-03-29 03:59:40.000000000","updated":"2026-04-30 11:53:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":961092,"cherry_pick_of_patch_set":4,"meta_rev_id":"ac874badc52e0124900487aace43ba0cc2830630","_number":982548,"virtual_id_number":982548,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-09 21:29:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-29 03:59:40.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"state":"CC"},{"updated":"2026-03-29 04:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 11:53:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5000624f52c22ee4fd471d107544bd1dbf90c8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 03:59:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1348d0ac685e69ff3749152ffa8525063da63760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 04:44:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b12a005e0f7c4524b246be5542ee0ce6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacb87b410124a99ac167a53e4ffac78 : FAILURE in 4m 42s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/82110567df044249a30b4e186c60e62f : SUCCESS in 10m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9d225cd504854497a795e688bce4e66c : SUCCESS in 7m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1cba829c887f4c48842e08f9b3bdf3b3 : SUCCESS in 4m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/74c896a0eec847b0b9e445cd1a478a2a : FAILURE in 37m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/62d7fb3e63764996b751f782472b54f7 : FAILURE in 20m 50s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8c06381d035e4757b528525d8fb99440 : FAILURE in 30m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9863742abc9147b98fdbe1aa64048685 : FAILURE in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4c99f65aa8b8e781664005dd6cf1bcf0eeb2ff71","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 09:09:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b1020ed95280bef94f5dbced2a3d035a890eb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 10:07:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56d5f38cc12a4ff9b3ca51a5a1386e86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b33ad6e0574d32b9bb186d216ee57f : FAILURE in 7m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ad95824618524f9ba5bf1e5b2105b268 : SUCCESS in 12m 02s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/af9ac7485b894374bb8e4962b7083529 : SUCCESS in 6m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/824817fe2a144fde869b783c01b26836 : SUCCESS in 4m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9ec3a65d8cb948169157615d2d9595d7 : FAILURE in 51m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9b0da232eeb54c64aa35e6ce936dabcd : FAILURE in 46m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7708525973794ffeaf030fea6e11fe77 : FAILURE in 19m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a5ba6eec939d4ccdb1acfac15da10d69 : FAILURE in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"69820be87c39ed46de12be24c1df32c5126a2a65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 20:26:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_30434\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"_revision_number":2},{"id":"87614c6fd108f822b9f0ecabf4e9acb7dd95eb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:29:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1444ea3b19bb46f78bf1143153c01960\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5977ed2fdd34d2ebd14ab47d8e00be3 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c14e1d81351941579db1f754a15fd3a3 : SUCCESS in 10m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ae5d281c38db49848b016d28db9a9812 : SUCCESS in 7m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c949ba905cef418d818cb139e8a50ddf : SUCCESS in 6m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/00f234797af0430d9c7dd211d7484518 : FAILURE in 55m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e9557656b7824558add592e3f4fc472c : FAILURE in 42m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/60999301a5ba4232bc0a0df01eb1ad59 : FAILURE in 32m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7c47082889a14a3580a03fd19d97cab1 : FAILURE in 30m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ac874badc52e0124900487aace43ba0cc2830630","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 11:53:50.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/skyline-console transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f03e1c4535627d4d3bb59e5d16b49080ac8a2005","revisions":{"fcc3efb69836b62505ca75082caf19732a009f16":{"kind":"REWORK","_number":1,"created":"2026-03-29 03:59:40.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/48/982548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/982548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/1"}}},"commit":{"parents":[{"commit":"da4d16212f1e6cbfbb2ddde68722f24383a6820c","subject":"Merge \"feat: update package for release\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/da4d16212f1e6cbfbb2ddde68722f24383a6820c"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-03-29 03:59:40.000000000","tz":0},"subject":"fix: invalid vars syntax in playbooks","message":"fix: invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n(cherry picked from commit 38fbc9258aaa8f06bbf07c120a336f209c68366f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcc3efb69836b62505ca75082caf19732a009f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcc3efb69836b62505ca75082caf19732a009f16"}]},"branch":"refs/heads/stable/2024.2"},"f03e1c4535627d4d3bb59e5d16b49080ac8a2005":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-09 20:26:47.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/48/982548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/982548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/982548/2"}}},"commit":{"parents":[{"commit":"7b25e4a526d4704013f3858ffcf7d3ebc827a0fe","subject":"fix(i18n): update locales and fill missing translations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b25e4a526d4704013f3858ffcf7d3ebc827a0fe"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2025-09-15 07:00:08.000000000","tz":330},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-09 20:26:47.000000000","tz":0},"subject":"fix: invalid vars syntax in playbooks","message":"fix: invalid vars syntax in playbooks\n\nChange-Id: Id7c86d1093314f6e4e8b9dd8b106df883b671ce0\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n(cherry picked from commit 38fbc9258aaa8f06bbf07c120a336f209c68366f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f03e1c4535627d4d3bb59e5d16b49080ac8a2005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f03e1c4535627d4d3bb59e5d16b49080ac8a2005"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
