)]}'
{"id":"openstack%2Fskyline-console~982552","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~I07fb09ccc2843cce2176fd86b342fe1c96550b02","project":"openstack/skyline-console","branch":"stable/2025.1","topic":"fix/setuptools-docs-2025.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-01 16:47:22.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I07fb09ccc2843cce2176fd86b342fe1c96550b02","subject":"fix: add setuptools for pkg_resources in doc build","status":"MERGED","created":"2026-03-29 09:48:41.000000000","updated":"2026-04-01 19:14:11.000000000","submitted":"2026-04-01 19:11:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982552-fix/setuptools-docs-2025.1","meta_rev_id":"72a6ba58418a2eee70bd3985f5064acc37829019","_number":982552,"virtual_id_number":982552,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 19:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-03-31 14:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-03-31 14:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-29 12:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85f11f2ba60d3559f069b1cacc6e2ddb4ca55485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 09:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf9a6688ba0163436318595c1707dd27d46de330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-29 10:00:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9991723ec0196767f1906ee04239f74ace436a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 12:16:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aed451d1e29b4612b99f5021b0abed77\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f624d4412abb411b912834c9b312328c : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831f571ce48f492b87a46701a5d5b9eb : SUCCESS in 5m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1cb56700bc54493499daded1d4d14a60 : SUCCESS in 11m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a83a96e608db488580295c2b4821f7bd : SUCCESS in 4m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1d54a7d214444dc5b13a0e11de930ebd : SUCCESS in 4m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a9ba3dd3bbdf4fd19675113dbe318ad4 : FAILURE in 2h 10m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/015073d37c3343b6a326c8bc27322a8b : FAILURE in 1h 36m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3514d281c94b4864a8e7060d5287c3ac : FAILURE in 57m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e23447e1d9ce4ca49445c8a7fc0607a2 : FAILURE in 1h 40m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ee9051a79e48815b21565803171121d645be19b1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-31 07:55:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69c44f4e85dd8724d1fbc4d9d649604dec5f22ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:28:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf95633041240e4a3072d980d85ef87\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/203173ce0fa141548de9814f9f67812f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1098031e2844ee970df9eff7697587 : SUCCESS in 3m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dc54308e38f841b783d35f90642fea92 : SUCCESS in 8m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/effce4f8de4b4850a7bb4f5e21e163f8 : SUCCESS in 5m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b48022195cef41fd9049e7828c7c1869 : SUCCESS in 5m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/852c563b1d404a6dbee62a7fae68329a : SUCCESS in 1h 32m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fc3178ae958746f1823da14c7149e621 : SUCCESS in 2h 27m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f323ffe8bc1d461fadbf8ef49f0e6479 : SUCCESS in 52m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/342c2e05830a4beca5f949b7affdba0d : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":2},{"id":"22c2f8c072f75f052d4a6e359233da3c41bae47d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-03-31 14:15:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6e51c0b837bb349809c840a0c64942fc1636c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:15:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9cc7a2f06fa9f230bd95f03a19c6459c17a928ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:27:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8f9645e109a4f3aae18e3827b772e68\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3dabc2993384752b2e3441eb1d0de7e : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267af2ec65aa4d36bde199a4c8583366 : SUCCESS in 7m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fbd6930ed21746e888092f51509e9f75 : SUCCESS in 9m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7d590535eb214527b52f48587b187585 : SUCCESS in 4m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/741e23402926446c91dc5379fc7e3328 : SUCCESS in 8m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/54babde013744694b76662053e9f3ddd : SUCCESS in 1h 23m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/948142b8232643f2a64c77b6e660c597 : TIMED_OUT in 3h 04m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9593bb0f137943b6ac62d7944c1d2d5a : SUCCESS in 55m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b4c4f55cb13c4d8eaf093c3047939e6c : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a8038828d2f94dd17bef48f1ef5b93b1937b4eed","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 05:51:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b950f6f4420ee45ec4e6cc09bf15982ec2ce198d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 07:37:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7857df72535641bea3ce60563445684b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd15b3c4869b479d94c8cedfc5307a7b : SUCCESS in 1m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28af843ca8984c0d8234e649024e8ec7 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cbfd60a6ff664a0ab2449585313deb5a : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b4b022393c4440e88e6401ed5ee7cdf4 : SUCCESS in 8m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d00f8f025ed44a418266f09998b540f4 : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6ed1d3e10adf4649a31a80c63592420e : FAILURE in 56m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/47a2176be08e44f99c5ca662be26d68f : SUCCESS in 1h 40m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7cc312cda2ce46a39d8edb00b057b182 : FAILURE in 58m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/276cd58b8370413d8a18567ca45d2882 : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0f753978a8958eb8d232f6d93f78b7cab8fb7f96","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 07:44:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bc8aa766bc601a06ecbb366d58a25ddf0e50dce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 09:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3310caa4b04f7485505f17f276ec6c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09b0a83ed8843598a75d64cdac3430c : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091a922237634835a4cb425583a04c9d : SUCCESS in 6m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6715df4dedb944d0bc293980f88bd1e4 : SUCCESS in 4m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/225ab1566036415f81dbcc41aca9c361 : SUCCESS in 6m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c1f0ae09f99a4057acedbe9b87e53613 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/021a5da315fe49f3929b7697a1d4fcba : SUCCESS in 1h 07m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/48119c834e2543e3bfe2229fb4164d6d : SUCCESS in 1h 54m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aded47664a7b4a0eb07e5fae398d42da : SUCCESS in 31m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/51ff03a636b14ea39b6455952492fecb : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f7f2dc53bce2e3889ba4892ef28f197621172484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 09:48:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c14fa301e45e3a5c441b8746c54983d46c76a7ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 11:59:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d57b4c81a8f4386bcc87e256b905ac8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca6820ec9059496991fcb5861f7b0ecd : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1fb00601f5e49c5a44340a68b87980d : SUCCESS in 3m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5275627a98ff4f48964a966ad8f5cc3d : SUCCESS in 12m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/453b1d6323104abab0ebbf55a3b9af00 : SUCCESS in 3m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2da855be6644400d849bef1341558452 : SUCCESS in 3m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a42c0e1da5e5455bbb9b8861ff763e89 : FAILURE in 53m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/aaeed2f3404a4c599c563a8f7db3493e : SUCCESS in 2h 06m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9a4e7fd8918a429a9084de2e5fdd5e93 : SUCCESS in 53m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/39a1c2fc3724450a8dbbbd6d69c08c7d : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a0cc83ffe2f20a7df36255e03cba5ff4acf09515","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 12:10:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61c078f7725485bad3104a69934340357d5568f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:45:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3755d4190fd840cdba19b8c205a59da0\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52caa735a9f64c428b0be1483d2c746a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363a312933544df694e26ca24bca8b0b : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b37f046af63c4fc5936a4e46486d1496 : SUCCESS in 8m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c5c002a232244b4facf38f5a03d85614 : SUCCESS in 5m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fdafec92cbd34602992ce6ee25c9000a : SUCCESS in 7m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/19a03877b72d4fb68f5a8f6b956cdfe6 : SUCCESS in 1h 28m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cf899ce57be446738cd068a03b24f0d1 : FAILURE in 54m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/42ed9f1346b74f7bac29b8663d7eb10a : SUCCESS in 46m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d5d211f72ba04eb6bd53894dddecd0fb : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":2},{"id":"daabffd517091625ec840b4e8c8b544fdaee8933","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 14:11:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"574718d3bb634f04e3faf416d7b9cef7b52256fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 15:52:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/847c41d16992445d8b4f8d385b32ae59\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/344994373432418891e611907b7e116c : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6e447bcca94ade9708a8691a0adcea : SUCCESS in 6m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5f501e70726144d1a30b558f8dfdeafc : SUCCESS in 11m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a5716c9b8c5425dbb943540b57d859d : SUCCESS in 5m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/806f89a85ffa4322b4cad3004e8cc952 : SUCCESS in 9m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1c4d1b32b56d4cbb9502cd4a42804eae : SUCCESS in 1h 35m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e6179e58f13a487e953837fb8a73f264 : FAILURE in 20m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d31c775cc7bb49ea82cbd4773ff8ec35 : FAILURE in 43m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c99c0d90006149e7b8074a6d26cf8a57 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d9cbf4eac96a912115233bd2e2e227b261ad8b13","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-01 16:47:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84b5aba776dfb804881ad0facbd6478a4df2c110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:03:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b4865d75c04720a2a66df8248fa7e7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60232f666c94f57a59e37fe4f3454f6 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffedd52c51b4be0993bd7a01d851274 : SUCCESS in 3m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/72a862a4ff704448bd356e4bce634237 : SUCCESS in 12m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/193cc85222ec449aac925b79e1531309 : SUCCESS in 6m 59s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/247a29d2a4624972b2e6c1a23447ab44 : SUCCESS in 4m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9d1a889ad7d2451db1a48dd5348c9f81 : SUCCESS in 1h 05m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f878d07acff148be950406ea4cf5e10b : SUCCESS in 1h 09m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b6539c2146844bd2add94876a9d18fee : SUCCESS in 57m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0b04a0151f2242abbe8dff1ab5ce3fd8 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":2},{"id":"187e57074bef329eedbaaebebd4e59645b672025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 18:03:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2315650494192928c7306863b489443a8e926572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 19:11:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d9705d1b0447e18b5396caf6115dda\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9d9c0288605488ebc67d7a6f5aea2f4 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3739cf1f40444e39a3e992179efa644b : SUCCESS in 3m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a3fbbbfbc02e48c39df6d0ff31216b25 : SUCCESS in 7m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/96eeb0cfc3a34f48ae1a9e7720bbfd89 : SUCCESS in 8m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8176169fd16249d38164355a1191a127 : SUCCESS in 4m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/348733c0d9ba418da88c5c6bc8bb3a4c : SUCCESS in 1h 05m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/61cca1d42b414f67bee16de455b71a93 : SUCCESS in 1h 02m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2404b8f45ff3429da33c67c50fa2dac0 : SUCCESS in 32m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c9d4f4c767714a118cf3401472156422 : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":2},{"id":"13491e6de11f99c520cefabd44f0e72341622256","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 19:11:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72a6ba58418a2eee70bd3985f5064acc37829019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 19:14:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9078e110cacf46f594d47293b95b910c\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e6dd19c74104c0699fb5e832d3a4df6 : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1358d06bb5e64465abba2d955fcc40dd : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82bc481f01330d4f50c2e5a60e8a3d86cd5adda5","revisions":{"e1a37afd7d4a4dbc15c34a48f90ba3e3e258f1e9":{"kind":"REWORK","_number":1,"created":"2026-03-29 09:48:41.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/52/982552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/52/982552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/1"}}},"commit":{"parents":[{"commit":"46fc2803ed3d895421adf45c1b719bad0dd930c7","subject":"Fix IPv6 subnet fields visibility in network form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46fc2803ed3d895421adf45c1b719bad0dd930c7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-29 09:22:53.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-29 09:22:53.000000000","tz":480},"subject":"fix: add setuptools for pkg_resources in doc build","message":"fix: add setuptools for pkg_resources in doc build\n\nPython 3.12 no longer includes setuptools by default in virtual\nenvironments. openstackdocstheme requires pkg_resources from setuptools\nto work properly.\n\nChange-Id: I07fb09ccc2843cce2176fd86b342fe1c96550b02\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a37afd7d4a4dbc15c34a48f90ba3e3e258f1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a37afd7d4a4dbc15c34a48f90ba3e3e258f1e9"}]},"branch":"refs/heads/stable/2025.1"},"82bc481f01330d4f50c2e5a60e8a3d86cd5adda5":{"kind":"REWORK","_number":2,"created":"2026-03-29 10:00:16.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/52/982552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/52/982552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/52/982552/2"}}},"commit":{"parents":[{"commit":"46fc2803ed3d895421adf45c1b719bad0dd930c7","subject":"Fix IPv6 subnet fields visibility in network form","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46fc2803ed3d895421adf45c1b719bad0dd930c7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-29 09:22:53.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-29 09:58:17.000000000","tz":480},"subject":"fix: add setuptools for pkg_resources in doc build","message":"fix: add setuptools for pkg_resources in doc build\n\nPython 3.12 no longer includes setuptools by default in virtual\nenvironments. openstackdocstheme requires pkg_resources from setuptools\nto work properly.\n\nChange-Id: I07fb09ccc2843cce2176fd86b342fe1c96550b02\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82bc481f01330d4f50c2e5a60e8a3d86cd5adda5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82bc481f01330d4f50c2e5a60e8a3d86cd5adda5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
