)]}'
{"id":"openstack%2Fskyline-console~983197","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~I9f5cc3e462c4819976fd2e23570d626060db0b49","project":"openstack/skyline-console","branch":"stable/2025.1","topic":"2025.1.ironic","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-04-04 13:28:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I9f5cc3e462c4819976fd2e23570d626060db0b49","subject":"test: Disable ironic deployment in DevStack E2E tests","status":"MERGED","created":"2026-04-02 12:51:29.000000000","updated":"2026-04-04 14:47:01.000000000","submitted":"2026-04-04 14:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983197-2025.1.ironic","meta_rev_id":"90d57ab1c0dbc05f84fae96086a1a72df6bb5a6a","_number":983197,"virtual_id_number":983197,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-04 14:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-04 13:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-03 21:18:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-04 13:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 13:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-03 21:18:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"1cef955e1c1a7aee8482a715afee9752795ab379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-02 12:51:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92a98c855e60624aa9cf433305ff56045b7f72a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 13:54:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/879c2981fe6c43b0a3f6b9afbd0dd952\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e57cb4fb32d42a68362b585c037bc8f : SUCCESS in 5m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/72a230af6b2a4006b3d212da7d1aa519 : SUCCESS in 4m 25s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ff5efeaaba14450f902b8bc0855dabec : SUCCESS in 4m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8d634d0bf47949b8ae1f7c0879346cd5 : SUCCESS in 3m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/42215d72a6ab435dadaaa0f270dd1ebd : SUCCESS in 1h 01m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7b53546afae2484690db112830c129c2 : FAILURE in 22m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/75cdd3826cc84d99b817ce95b7cd216d : SUCCESS in 57m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cbb516a0aae14b8589cf65300e77e541 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"06253ccdeca66443ef394ba138b9b90b6c1430b8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-02 15:09:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c46870ceaf8b581be28245a12e35bc7f15f00d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:22:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ccebee495a54ca78c0b0c33bc5ce006\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f453363d89420492f681cbd9996ce7 : SUCCESS in 7m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/490a0f50dcd34eed99fa585d3f4d6dc8 : SUCCESS in 7m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2a9afcd8264c48e8967a2ced745d9651 : SUCCESS in 8m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/470c2d96c63f416ba747697f22e23128 : SUCCESS in 5m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2804e72f32924b03b46fbecd05141b72 : SUCCESS in 55m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b359874e44be4d25a99c72badf459950 : SUCCESS in 1h 03m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/24c73ac02cd74e998397d74c410fcc32 : FAILURE in 1h 09m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9d4852cf3009450d85100da7c4c7cfc6 : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"56d87e9b15b8109d58dd42b3e0e1f244f3f4f7d6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-03 04:01:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d94da4883407b6902d38986ad02b6e736df1ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 04:46:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8e3f91b77d14389a90405da9b16e379\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eba051faccc4ca1b4365426a7fe3613 : SUCCESS in 5m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c93abb6f1020497cb40827f1e37f190e : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b22692cd3b56464e9e9f3351843a66d0 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/43636a7252ca499a93282e742478cc89 : SUCCESS in 8m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0f29901a61c34fa3a5c6fab7addae369 : SUCCESS in 43m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/530fb0e11b674ad59814c5ff1a4d8408 : FAILURE in 34m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/de3f6355a9564369b9031d086add35c8 : SUCCESS in 39m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6ce329fb4ed34869bc3ee639a41169ce : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"61633a8b79aec58167d731f01f5ad20bd897583f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-03 21:18:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc9f20a17fb1bb8b6f5455db55b953bed3aff7de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 22:50:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44dc4c6e5afe49dd97cb658ec05f5786\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce2e66c3c9d47c09f3a5faa70ad5f70 : SUCCESS in 4m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2d68d589b4524c7281e4b28aff2c7271 : SUCCESS in 5m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2714ed8b38a9494d8d06b2674471292e : SUCCESS in 7m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/41733bcf73834aceb773ee6d67d58b04 : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/60d69a7d406c49dca3fb2d700083417a : SUCCESS in 1h 25m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4c80837d4a2a4fdf8084fcfb570618f1 : SUCCESS in 1h 03m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/464e8281bbfc4a73879ed7e150f5a305 : SUCCESS in 53m 37s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/da4f2c9a697745ce8cd97e96b269a9ea : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3e950087bf5d5ab8102187cd49ecbec4a2ab3fa6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-04 13:28:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"060aafc698aef6df966cf212057b06d11cd35dbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 13:29:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b000888fe98e154dd291c31b208dd29ee51fc24f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 14:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ef125950f84aa48597f76ebe5f4d13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783400002b864e52aa2d58130e50b8bd : SUCCESS in 6m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a730d79e2a994e1d8d83f09053681804 : SUCCESS in 10m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8a294d475bbb4e5ab715abecd18e4f07 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/df04fc2ac408453bb1b52f2f72bf1033 : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2edd27ce2e60480d9bf745f806460d7c : SUCCESS in 1h 11m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b030a63643724a44b157dd3478344c57 : SUCCESS in 1h 02m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/79eb9c0ebfca4995a72418077291a723 : SUCCESS in 53m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8dfc22a9203b40338fd260e05f94086a : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cb4e097671bd7cb17c9481f068150a6565096501","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 14:45:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90d57ab1c0dbc05f84fae96086a1a72df6bb5a6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 14:47:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b492039dac964a15bf822eea3d1e0753\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f981b5d3f2d6470e93660be9713a1ff1 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c1e1d979912d2b3f1b8ad0b48264277132bd14a","revisions":{"6c1e1d979912d2b3f1b8ad0b48264277132bd14a":{"kind":"REWORK","_number":1,"created":"2026-04-02 12:51:29.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/97/983197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/983197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/983197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/983197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/983197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/983197/1"}}},"commit":{"parents":[{"commit":"82bc481f01330d4f50c2e5a60e8a3d86cd5adda5","subject":"fix: add setuptools for pkg_resources in doc build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82bc481f01330d4f50c2e5a60e8a3d86cd5adda5"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-04-02 10:17:16.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-04-02 12:50:58.000000000","tz":480},"subject":"test: Disable ironic deployment in DevStack E2E tests","message":"test: Disable ironic deployment in DevStack E2E tests\n\nDisable ironic deployment and skip ironic-related E2E test cases.\n\nChange-Id: I9f5cc3e462c4819976fd2e23570d626060db0b49\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c1e1d979912d2b3f1b8ad0b48264277132bd14a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6c1e1d979912d2b3f1b8ad0b48264277132bd14a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
