)]}'
{"id":"openstack%2Fskyline-console~983348","triplet_id":"openstack%2Fskyline-console~stable%2F2026.1~If961889314831eee90197a372a6b970b0af00796","project":"openstack/skyline-console","branch":"stable/2026.1","hashtags":[],"change_id":"If961889314831eee90197a372a6b970b0af00796","subject":"fix(i18n): update locales and fill missing translations","status":"MERGED","created":"2026-04-04 13:29:16.000000000","updated":"2026-04-04 15:01:45.000000000","submitted":"2026-04-04 15:00:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983348","cherry_pick_of_change":982360,"cherry_pick_of_patch_set":1,"meta_rev_id":"cf4db8dac35868764aa81336179d03416764922e","_number":983348,"virtual_id_number":983348,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-04 15:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-04 14:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-04 14:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-04 13:44:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e5c090991ac30b1526dfbc95df819430515f34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-04 13:29:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3861215923405b74f489ec240f291c45a879efad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 13:44:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee6ad3d6e0554401bf0710a8e0fb63aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a14b298f1274c30bbf6803b45faed08 : SUCCESS in 7m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a41935164d0a4a2e8c34ae114d7cae86 : SUCCESS in 4m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/068993276d6741b5b191c80304a7a44a : SUCCESS in 9m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/70a05e450fe94c26a42d2ae021572404 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"37fa6272fba8fe1d3abe63782ff8a6a8ec5f2ad5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-04 14:37:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b71da7e6c25431f88a78a936900e34ddee6be0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 14:37:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bf95946708d177e2875457fedf37db2e07481a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 15:00:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c8ae4f2a41410aba644f0c06e2b46e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7bd84a15814dd08582b8348e1afa48 : SUCCESS in 6m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bfa20c69771a49bdbe657f6cf7c3261d : SUCCESS in 8m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0b3bd70202e14951a3483390c4d407b4 : SUCCESS in 8m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/89c945dd9170497f9f0d7233cae22847 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fa88b2d3a5b0db5144d0fd9f4795bc20ae485156","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 15:00:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf4db8dac35868764aa81336179d03416764922e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-04 15:01:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd13e3551e0e42428c9d26940a71ff8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae91359c07e41c09529aeb285100fb8 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90e8e4adbb2f2cc157ee405dbf71d11f30f367c1","revisions":{"90e8e4adbb2f2cc157ee405dbf71d11f30f367c1":{"kind":"REWORK","_number":1,"created":"2026-04-04 13:29:16.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/48/983348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/983348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/983348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/983348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/983348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/983348/1"}}},"commit":{"parents":[{"commit":"de6f92d7a6a10626c5d1eba63ec35e804e61e85d","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2026.1\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/de6f92d7a6a10626c5d1eba63ec35e804e61e85d"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-03-27 06:31:47.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","date":"2026-04-04 13:29:16.000000000","tz":0},"subject":"fix(i18n): update locales and fill missing translations","message":"fix(i18n): update locales and fill missing translations\n\n- Add new string \u0027Any Availability Zone\u0027 to en.json\n- Fix case for \u0027release fixed IP\u0027 in en.json\n- Fill missing translations for zh-hans, ko-kr, tr-tr\n\nChange-Id: If961889314831eee90197a372a6b970b0af00796\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n(cherry picked from commit 73db764e8d43c23ee0694f0d46e1b37dc457f1ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/90e8e4adbb2f2cc157ee405dbf71d11f30f367c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/90e8e4adbb2f2cc157ee405dbf71d11f30f367c1"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
