)]}'
{"id":"openstack%2Fskyline-console~983500","triplet_id":"openstack%2Fskyline-console~stable%2F2024.2~I04a4690d1edf2eca2091e6fbd00e7ca5a5fc5bcc","project":"openstack/skyline-console","branch":"stable/2024.2","topic":"2024.2.devstack","hashtags":[],"change_id":"I04a4690d1edf2eca2091e6fbd00e7ca5a5fc5bcc","subject":"fix: add setuptools for pkg_resources in doc build","status":"MERGED","created":"2026-04-07 03:23:13.000000000","updated":"2026-04-07 04:07:51.000000000","submitted":"2026-04-07 04:05:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983500-2024.2.devstack","meta_rev_id":"c1f75cc4442fe56fc9dfdef1a6247c49389bbb20","_number":983500,"virtual_id_number":983500,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 04:05:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-04-07 03:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-04-07 03:53:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 03:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19da4caae2ebdd08878f0e67d829396f73e5ea83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-07 03:23:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59a47dc6d3ba6b344f613dad301190e8cb577cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 03:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e36b564bd72541728fc14c3819a0d019\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632c7c93400a4868a8f3bf8b08e1a158 : SUCCESS in 1m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc6c5439c4d4890b62ec4483a767a00 : SUCCESS in 3m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/47003fa1fe7d4a039e74ab744895f5b5 : SUCCESS in 5m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cb200e24559e4934b65cea52dde357b7 : SUCCESS in 4m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/13f00cec3c3b4ca3a188b93a35b519cf : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d1a00fe8b45ba04b0345686cdec54429ccaf2f70","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-04-07 03:53:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"985e7452a151151a7aeab58e5676293ceb73c6a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 03:54:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc353722ca5e39acb2e1d0fdc1e74c8945630e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 04:05:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60556538256405f8b5d855e3842858c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd1e62cb5d2d46998d83c8dc9a583814 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c30fd41ecb4a28af97d0c773107c96 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f0ec03df21f3432099d092e7fafac396 : SUCCESS in 4m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f7c2e534a8c74f2bb28804e58ff4a041 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ca6eac20a9f848369b57917d7c338073 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7509a00ddc448008c95762b1c828113d6be05a6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 04:05:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1f75cc4442fe56fc9dfdef1a6247c49389bbb20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 04:07:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40bba5d98a49447dbb2065b3ff513785\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b607826ce1a45c68624f57658f2c7fb : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c675d7ad5a9f4e26827814969137135a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2532d409941df8dc9d44853d11376e072db727d","revisions":{"c2532d409941df8dc9d44853d11376e072db727d":{"kind":"REWORK","_number":1,"created":"2026-04-07 03:23:13.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/00/983500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/00/983500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/983500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/983500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/00/983500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/00/983500/1"}}},"commit":{"parents":[{"commit":"da4d16212f1e6cbfbb2ddde68722f24383a6820c","subject":"Merge \"feat: update package for release\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/da4d16212f1e6cbfbb2ddde68722f24383a6820c"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-04-07 03:20:48.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-04-07 03:21:23.000000000","tz":480},"subject":"fix: add setuptools for pkg_resources in doc build","message":"fix: add setuptools for pkg_resources in doc build\n\nPython 3.12 no longer includes setuptools by default in virtual\nenvironments. openstackdocstheme requires pkg_resources from setuptools\nto work properly.\n\nAlso add some removal files for e2e cases\n\nChange-Id: I04a4690d1edf2eca2091e6fbd00e7ca5a5fc5bcc\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2532d409941df8dc9d44853d11376e072db727d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2532d409941df8dc9d44853d11376e072db727d"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
