)]}'
{"id":"openstack%2Fskyline-console~985411","triplet_id":"openstack%2Fskyline-console~master~Ic433f81c7a8d04df8ebb9796b82bbe14807e5167","project":"openstack/skyline-console","branch":"master","topic":"send_first","attention_set":{"5575":{"account":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"last_update":"2026-04-20 14:09:05.000000000","reason":"\u003cGERRIT_ACCOUNT_5575\u003e replied on the change","reason_account":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"}},"38847":{"account":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"last_update":"2026-05-18 02:21:46.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic433f81c7a8d04df8ebb9796b82bbe14807e5167","subject":"Allow to specify metadata options on instance creation","status":"NEW","created":"2026-04-20 13:34:43.000000000","updated":"2026-05-18 02:33:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"843e9f816a5fc119b3d162fba947391a375b7418","_number":985411,"virtual_id_number":985411,"owner":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-28 10:49:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":-1,"date":"2026-05-18 02:33:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":-1,"date":"2026-05-18 02:33:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 14:09:05.000000000","updated_by":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"reviewer":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"state":"REVIEWER"},{"updated":"2026-04-20 16:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-18 02:21:46.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"9200076db82718803d9d7a8965e48e4d841db8a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-04-20 13:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e89ca18ef932ea081edf4c9b50f265c8c15c0b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 16:06:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/652de3a49dc34556bc24adf89edd46ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7baa2953624c0abd148742c46c119d : SUCCESS in 8m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8e8b5a5d852c45fcba288c1cfd362648 : SUCCESS in 12m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/72c2fe6c2b6c45dc91c407bee8716922 : SUCCESS in 9m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c0981d8b812a4c5ead1326900b01cc10 : SUCCESS in 9m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/115d4182939c44238302fc1582a8e316 : FAILURE in 1h 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6a18312c43e3c7cb84ecfbbf6fc2d15331905d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-04-23 14:29:36.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"883b0094a59631765ca2048487ca3ed7e336ad9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 17:05:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec5bb2d497cf4c388a37f31b7746f659\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37336c2bc9c8479c89fffa87176976f1 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8df8088d8ca146bcbef902b3f17f133b : SUCCESS in 11m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3c353942dd274b96b3daa26b7d28d115 : SUCCESS in 7m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/922bc122813d48af9a2455dbaeda5f4b : SUCCESS in 7m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a901338f89bf47119e0a9f194c07c097 : FAILURE in 2h 25m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d7028d2890f7185917e56c9b6cac9b41b5051a60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-04-27 09:05:31.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a90c92937e3cdf0c26daeeac5bce6a6d2c04faa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:40:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63dde1f51a704d31beccdbb82f0b332f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e021efe28a064647aafa60cad897d7ce : SUCCESS in 6m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/53d43649cb8b406988ba38976aeb71f7 : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ec3a6965d6a544d0bf7d251b60120434 : SUCCESS in 5m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/89ea460335df4695b375ad52064fadba : SUCCESS in 3m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/107d4dd320fd4a87863804320774ae79 : FAILURE in 1h 28m 59s","accounts_in_message":[],"_revision_number":3},{"id":"c8e73e58ca4a8e3eaaf13dbf1c02fe088a0d545e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-04-27 11:30:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8310b416acb8721a6c41d863cc7fee7782e31b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 12:35:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d97c7adfc0d43e898e4ddfecea0fd3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cab3ddd0274e9cb0fe5048eb98c836 : SUCCESS in 7m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/edf46bfd4440432591c351fc3aa88d2f : SUCCESS in 8m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d94e66cd38294707a991d0c830364090 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dd3ac046bef0440cb90c49c41f175aae : SUCCESS in 6m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9a012bcc9f2644f4b7a1fdf35b8ad70c : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":4},{"id":"375de20e6036de07920222c17e7671031cfec649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"date":"2026-04-28 10:02:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7586fba90959bc00b4999e56539b608be0016816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 10:49:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1a87cf5ff2444b958213cee1ac7bc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b90d3f9d6cc42e8a19b45fd4e96dd98 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2e05e3bfd30c4773ab01ff3373106376 : SUCCESS in 4m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4bf50ba71dc14b54b92b7cff543909d6 : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ecbc43695dce48c0a826156343caafc5 : SUCCESS in 4m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9c4e1e00d7774c8db5a5e64470dcb40a : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":5},{"id":"03a18352dbb89df3fa24341b25b8d8257fc407c4","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-18 02:21:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a8b335edeea285ef03185394dbdebc9f8bb21d4e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-18 02:21:53.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"843e9f816a5fc119b3d162fba947391a375b7418","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-18 02:33:22.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f3d00a394f50f8749acfcb3f608e86159100d67b","revisions":{"65cc6dc157c68e453d019977f362e3d5de385469":{"kind":"REWORK","_number":1,"created":"2026-04-20 13:34:43.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/11/985411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/11/985411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/1"}}},"commit":{"parents":[{"commit":"737b8f2ff06da8825f07ede2084f21a291c86f9b","subject":"Add region selector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/737b8f2ff06da8825f07ede2084f21a291c86f9b"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-03-19 13:50:34.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-04-20 13:35:04.000000000","tz":120},"subject":"Allow to specify metadata options on instance creation","message":"Allow to specify metadata options on instance creation\n\nAdded the ability metadata key-value pairs when creating an instance in the Skyline console.\n\nCloses-Bug: #2148894\nChange-Id: Ic433f81c7a8d04df8ebb9796b82bbe14807e5167\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/65cc6dc157c68e453d019977f362e3d5de385469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/65cc6dc157c68e453d019977f362e3d5de385469"}]},"branch":"refs/heads/master"},"5b61d48f643f35d6a780dc428585224520c8bbce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-23 14:29:36.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/11/985411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/11/985411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/2"}}},"commit":{"parents":[{"commit":"737b8f2ff06da8825f07ede2084f21a291c86f9b","subject":"Add region selector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/737b8f2ff06da8825f07ede2084f21a291c86f9b"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-03-19 13:50:34.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-04-23 13:04:00.000000000","tz":120},"subject":"Allow to specify metadata options on instance creation","message":"Allow to specify metadata options on instance creation\n\nAdded the ability metadata key-value pairs when creating an instance in the Skyline console.\nDepends-On: I45937184e7d87d6d7f4bd2a77cf0e72aff145aa3\nCloses-Bug: #2148894\nChange-Id: Ic433f81c7a8d04df8ebb9796b82bbe14807e5167\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5b61d48f643f35d6a780dc428585224520c8bbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5b61d48f643f35d6a780dc428585224520c8bbce"}]},"branch":"refs/heads/master"},"8db585228f0b766f3750f8af8300c2c7a480c0a0":{"kind":"NO_CHANGE","_number":3,"created":"2026-04-27 09:05:31.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/11/985411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/11/985411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/3"}}},"commit":{"parents":[{"commit":"737b8f2ff06da8825f07ede2084f21a291c86f9b","subject":"Add region selector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/737b8f2ff06da8825f07ede2084f21a291c86f9b"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-03-19 13:50:34.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-04-27 09:04:32.000000000","tz":120},"subject":"Allow to specify metadata options on instance creation","message":"Allow to specify metadata options on instance creation\n\nAdded the ability metadata key-value pairs when creating an instance in the Skyline console.\nDepends-On: I45937184e7d87d6d7f4bd2a77cf0e72aff145aa3\nCloses-Bug: #2148894\nChange-Id: Ic433f81c7a8d04df8ebb9796b82bbe14807e5167\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8db585228f0b766f3750f8af8300c2c7a480c0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8db585228f0b766f3750f8af8300c2c7a480c0a0"}]},"branch":"refs/heads/master"},"87580e22da936524b3fda292c1ecda2e5bb142ad":{"kind":"REWORK","_number":4,"created":"2026-04-27 11:30:21.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/11/985411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/11/985411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/4"}}},"commit":{"parents":[{"commit":"737b8f2ff06da8825f07ede2084f21a291c86f9b","subject":"Add region selector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/737b8f2ff06da8825f07ede2084f21a291c86f9b"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-03-19 13:50:34.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-04-27 11:27:24.000000000","tz":120},"subject":"Allow to specify metadata options on instance creation","message":"Allow to specify metadata options on instance creation\n\nAdded the ability metadata key-value pairs when creating an instance in the Skyline console.\nDepends-On: I45937184e7d87d6d7f4bd2a77cf0e72aff145aa3\nCloses-Bug: #2148894\nChange-Id: Ic433f81c7a8d04df8ebb9796b82bbe14807e5167\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/87580e22da936524b3fda292c1ecda2e5bb142ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/87580e22da936524b3fda292c1ecda2e5bb142ad"}]},"branch":"refs/heads/master"},"f3d00a394f50f8749acfcb3f608e86159100d67b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-28 10:02:27.000000000","uploader":{"_account_id":38847,"name":"Nursena Bitirgen","display_name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","username":"senabibi"},"ref":"refs/changes/11/985411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/11/985411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/11/985411/5"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-03-19 13:50:34.000000000","tz":60},"committer":{"name":"Nursena Bitirgen","email":"nursena.bitirgen@cern.ch","date":"2026-04-28 09:44:42.000000000","tz":120},"subject":"Allow to specify metadata options on instance creation","message":"Allow to specify metadata options on instance creation\n\nAdded the ability metadata key-value pairs when creating an instance in the Skyline console.\nDepends-On: I45937184e7d87d6d7f4bd2a77cf0e72aff145aa3\nCloses-Bug: #2148894\nChange-Id: Ic433f81c7a8d04df8ebb9796b82bbe14807e5167\nSigned-off-by: Nursena Bitirgen \u003cnursena.bitirgen@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3d00a394f50f8749acfcb3f608e86159100d67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3d00a394f50f8749acfcb3f608e86159100d67b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
