)]}'
{"id":"openstack%2Fskyline-console~987804","triplet_id":"openstack%2Fskyline-console~master~I5bf56c2627a94628bbc186dca412a45db735306f","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-05-26 22:25:10.000000000","reason":"Change was submitted"},"38431":{"account":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"last_update":"2026-05-18 05:34:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38431\u003e replied on the change","reason_account":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"}}},"hashtags":[],"change_id":"I5bf56c2627a94628bbc186dca412a45db735306f","subject":"Set browser title from current page","status":"MERGED","created":"2026-05-08 07:04:46.000000000","updated":"2026-05-26 22:26:18.000000000","submitted":"2026-05-26 22:25:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"987804","meta_rev_id":"8788787fcec694df0b314f8e5144bee2931b7089","_number":987804,"virtual_id_number":987804,"owner":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-26 22:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-05-17 01:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-05-18 02:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-11 20:49:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-05-18 02:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 08:10:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-08 09:12:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-11 20:49:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-11 20:49:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-05-17 01:17:49.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"177ffce1ff534ac8f74f9a179e8e380ffe9e1b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"date":"2026-05-08 07:04:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"983222332ff1488dae3b6cd3d8a8d9b0078829ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"date":"2026-05-08 07:05:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9eca5e59970f08ec59c819b3acdc384fe1149670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 08:10:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d62cb923d3489695c4d69bbf4bf551\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62e0dba109c54b5290963902c8fede3a : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec59d8197e04428b4e360e20509ecaa : SUCCESS in 6m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6e04ecf9ce914382ab4fa810eaa664dc : SUCCESS in 12m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/85b809f681e245b4bb90ba8dfc721cdc : SUCCESS in 6m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5e16d7bfca424d2481314d88f5fe76b7 : SUCCESS in 8m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0ee15e8698fc47278be847ab6404b722 : SUCCESS in 58m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a8b3bbb1714c4bd581f65741d16c4e9d : SUCCESS in 59m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/266de257de02490a8d0ee4e9d6f1b4a0 : SUCCESS in 54m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/56daf7c804bd42b6ae85ed34e4e43cfb : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1a26a9f579d6c55e5317e440eba08ea01eef64bb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 09:12:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b521a22a375f40672f0420b96a8aae2195e87a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"date":"2026-05-09 13:45:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"21bafde33b85880dcb11ac67c742559c274f8fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 15:04:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1868528cb4974682b12df139800b49ed\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b2a68c3481b4ec9b90eda5ba9f8165b : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8b730d575d43d88fdd282e7b7c3669 : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2e3e97836d704b50adfc753004c5dcf5 : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/acee9194a83b4eb893f5bd86b4aed211 : SUCCESS in 7m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6de02f4c6fd4445cb95e749a1e663622 : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1ec7a4ad74e74649ba494139e90d131c : SUCCESS in 1h 12m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2c8f687668c3412da0757fcffb700811 : SUCCESS in 33m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f2f8576ec61f4279a969e4f0f557c71e : SUCCESS in 52m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7cdffb8df8174c119b0a122cfa245c1e : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":3},{"id":"90c945929d63c5bb4ddd5c43205b3a2a7afbd73f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 20:49:17.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2847d1fbb46a54ad4cc8446dc40bcde160ded32c","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-17 01:17:49.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bbaa0423b31bac6763f7bb7d699722e2d2e3a29c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-18 02:23:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"750c6898c6c34db63e5af787b1d7398f04de2d3e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-18 02:23:24.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec8c7c4c0e50dd016b40da36ef46ff372421a926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 02:23:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fbde3c7ba452cfb8a3570eaed7270d0a9dc9bbf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 03:38:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69ee565b12f64121ae38261120fa3d3c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a37004bba344fc8aa748437dd20cf57 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6892584937334e3f81a4ce8dc1733bb2 : SUCCESS in 7m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c13980d1b027418f9b1716a3c45dcdb4 : SUCCESS in 9m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2ebb34cc850b49df8378476a91f9eea5 : SUCCESS in 4m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c03f71c47e584572aa853d4282f44859 : SUCCESS in 4m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/788d8a8654554d48b6719c3e93d01e8b : SUCCESS in 42m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/00ec75d8e573425299d5685a514874e9 : SUCCESS in 43m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b4bb9af81a4b48218385790bb9fa42b8 : FAILURE in 1h 09m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ef3f1e00bcc44ccab5949b058453f8f9 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":3},{"id":"849651e33a657acf9a689325b486e2ea6648f862","author":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"date":"2026-05-18 04:00:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4324989c59d427cf0ae846b8f39084e58d811e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 04:52:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3c7b892c1e44c2ca6dfaeeab9e350da\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d350a3ed278c4166a6fbd6d01b384b3c : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68affae7227b421b8b202f00b0df84a1 : SUCCESS in 6m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/90e9f9dfa96d442caed8b2005770d082 : SUCCESS in 10m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d4b2a5af1ff2458995ba3418ac39992e : SUCCESS in 7m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/78ba559b453a4e7ea8d0b58b23e73915 : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6c76f837e93a409696edafc8f2b75550 : SUCCESS in 43m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3562c2cb7b914ebfb7f38653e7367d12 : SUCCESS in 45m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/04cc7f482a6a4358bdcace5687a6e2fc : FAILURE in 51m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/080a935a472c4550b1f322bdadd1554d : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":3},{"id":"76fd7738747feb52420e7081ef48d28120eb8791","author":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"date":"2026-05-18 05:43:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6ab1b9ff90c1a7a379bfb77dd6c53711f4e45149","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-26 19:47:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"993321eafa4005ec1c0353340443fdc1cde7d9b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 21:10:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf239df94eb4ff1a59b7c586a0d7b70\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e724cbe0c493497f991bc9ed01d228ae : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f9836fc41f42a8903fbe57b72ff664 : SUCCESS in 5m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/053f6108889a4985a2ebe172169a35ad : SUCCESS in 5m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7539e0e367244bdd8055451281344390 : SUCCESS in 9m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a69f22456dec4daa8d2464156e38e02d : SUCCESS in 5m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a7f783412cda42e6b9eae2679aa8599b : SUCCESS in 1h 10m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c1a2b2a08a4f428c94158d074574ae08 : SUCCESS in 1h 01m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6edfafd30c1847369f473e29131f4e98 : SUCCESS in 52m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3369ec08ad98451c85d4317c5ac67e37 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b5161cbe52df4b414ed9ed4df2ad295ae12c177a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 21:11:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bf4eeb83cc4fbff12a86542e2128cf6dd8967638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 22:25:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f440fdbcfeed4d4885473fba58fa20ea\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd9482f4f5ae4faf8e0b71346fa9af1a : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbe7aa60b284d4889067c7df6635e52 : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/958558f7ea8945038b9222313bece30f : SUCCESS in 10m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0733fb725d584bd9b3d4a6f8210230cb : SUCCESS in 5m 55s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2bc4e33162994c38ae784243bc025b3b : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7eb5e037ede14d858d16b18fd710f540 : SUCCESS in 1h 09m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6b976e3950214084a3c1875760fa9fcc : SUCCESS in 30m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f3d630d2346946289fffa5ecede25f38 : SUCCESS in 45m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/20a0110e827d414db78ead9d8f700be0 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":3},{"id":"5eb64654f735634e9b079ec7c71e106e52961bae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 22:25:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8788787fcec694df0b314f8e5144bee2931b7089","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 22:26:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16d027e8a1b411c86a29acb42efe47d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50dc288e00f4c09bf6a7823438d2907 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec07433be494afc94e308b5ea52a912 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3b017fe0f2777570c72d89e62615a899fa56a267","revisions":{"7775279547fdc4a1670a890a14fbdcd258633033":{"kind":"REWORK","_number":1,"created":"2026-05-08 07:04:46.000000000","uploader":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"ref":"refs/changes/04/987804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/987804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/1"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-08 07:01:51.000000000","tz":540},"committer":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-08 07:04:30.000000000","tz":540},"subject":"Set browser title from current page","message":"Set browser title from current page\n\nSkyline Console currently keeps the browser tab title as \"Cloud\" across pages. This makes it hard to identify the current page when multiple browser tabs are open.\n\nThis change derives the document title from the matched menu route and uses the configured site title as the fallback. For example, the Compute \u003e Instances page now displays \"Instances - Compute - Cloud\".\n\nCloses-Bug: #2151848\nSigned-off-by: jaehanbyun \u003cawbrg789@naver.com\u003e\nChange-Id: I5bf56c2627a94628bbc186dca412a45db735306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7775279547fdc4a1670a890a14fbdcd258633033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7775279547fdc4a1670a890a14fbdcd258633033"}]},"branch":"refs/heads/master"},"dda07b29b55f256249a32e216a6e8cc034efbcfd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-08 07:05:23.000000000","uploader":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"ref":"refs/changes/04/987804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/987804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/2"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-08 07:01:51.000000000","tz":540},"committer":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-08 07:05:03.000000000","tz":540},"subject":"Set browser title from current page","message":"Set browser title from current page\n\nSkyline Console currently keeps the browser tab title as \"Cloud\" across\npages. This makes it hard to identify the current page when multiple\nbrowser tabs are open.\n\nThis change derives the document title from the matched menu route and\nuses the configured site title as the fallback. For example, the\nCompute \u003e Instances page now displays \"Instances - Compute - Cloud\".\n\nCloses-Bug: #2151848\nSigned-off-by: jaehanbyun \u003cawbrg789@naver.com\u003e\nChange-Id: I5bf56c2627a94628bbc186dca412a45db735306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dda07b29b55f256249a32e216a6e8cc034efbcfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dda07b29b55f256249a32e216a6e8cc034efbcfd"}]},"branch":"refs/heads/master"},"3b017fe0f2777570c72d89e62615a899fa56a267":{"kind":"REWORK","_number":3,"created":"2026-05-09 13:45:51.000000000","uploader":{"_account_id":38431,"name":"jaehanbyun","display_name":"jaehanbyun","email":"awbrg789@naver.com","username":"jaehanbyun","status":"github.com/jaehanbyun"},"ref":"refs/changes/04/987804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/987804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/987804/3"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-08 07:01:51.000000000","tz":540},"committer":{"name":"jaehanbyun","email":"awbrg789@naver.com","date":"2026-05-09 13:45:27.000000000","tz":540},"subject":"Set browser title from current page","message":"Set browser title from current page\n\nSkyline Console currently keeps the browser tab title as \"Cloud\" across\npages. This makes it hard to identify the current page when multiple\nbrowser tabs are open.\n\nThis change derives the document title from the matched menu route in\nthe authenticated console layout and uses the configured site title as\nthe fallback. The blank layout keeps the existing title behavior so the\nlogin page can stay at its default title.\n\nCloses-Bug: #2151848\nSigned-off-by: jaehanbyun \u003cawbrg789@naver.com\u003e\nChange-Id: I5bf56c2627a94628bbc186dca412a45db735306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b017fe0f2777570c72d89e62615a899fa56a267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b017fe0f2777570c72d89e62615a899fa56a267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
