)]}'
{"id":"openstack%2Fskyline-console~989527","triplet_id":"openstack%2Fskyline-console~master~I8b0e3c81590a646954c4997ec227c05dfc2bd5cf","project":"openstack/skyline-console","branch":"master","topic":"bug/2153560","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-05-26 13:04:13.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"38396":{"account":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"last_update":"2026-05-26 14:21:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b0e3c81590a646954c4997ec227c05dfc2bd5cf","subject":"Fix magic input for only options-filters","status":"MERGED","created":"2026-05-21 12:50:59.000000000","updated":"2026-05-26 14:22:33.000000000","submitted":"2026-05-26 14:21:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989527-bug/2153560","meta_rev_id":"23603929031797e9588810a37e01273195025d97","_number":989527,"virtual_id_number":989527,"owner":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-26 14:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-05-26 13:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-05-26 13:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-21 14:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 03:34:53.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"5ddf1916115e1b057deaf638372ece10db0e98b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"date":"2026-05-21 12:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7720bec27ac0f5ec5e1c70912e309feae3195957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 14:05:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/799cc5cf5a804e15badf65fa0738a95e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d2c791e80f5494bb1ea12582ccffde8 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa4d892cc584281a997633de19c1f74 : SUCCESS in 4m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/af1e453ce68647a896071e3dd620cdd0 : SUCCESS in 10m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3ef22eaea4ea4db58572e6173ab1224f : SUCCESS in 7m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e268cd85c7094a2986565070040f61eb : SUCCESS in 3m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/890779bef20b41ccae8212b79c7d0e58 : SUCCESS in 55m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/af10cc1ec1bc4ed09cab8af9659f59ab : SUCCESS in 49m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9b4b9b070df24a27b8e495ace06a9ac3 : FAILURE in 1h 08m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/157c5f9b7e4f40469a2839e9578fdfbb : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b5274326b73dcabe589fc03cfd431ca6c2ac225b","author":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"date":"2026-05-22 11:11:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dab57a096363ca2feee61d4be3a88e2f03690dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f8344ae99642129775c84928eae30e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c976c8577e5440f7904c0ad4b5c5e6d4 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65081c7206bd4100903c5506b41b5197 : SUCCESS in 6m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8c0da66f4a9c47a197cb6173f26d32ce : SUCCESS in 7m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b75e267d62f84534ade68081cf696b23 : SUCCESS in 7m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2d1642d4ada042ff96248cabe3fe350f : SUCCESS in 8m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/647ef459107c47feb29743bd47277148 : SUCCESS in 1h 11m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/eb7e0a5ea1ff43089a3e37c61bcc8ee7 : SUCCESS in 30m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7bf4b6ff9912428ba627021d16308656 : SUCCESS in 56m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ce826eebedc540fe9ad1a958f90464ed : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"34e4ae81ef3082d101dbbc0f2b7bbc92e31deb7c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 03:34:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f46363eb480ded903b70409b924104027a9f212","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 03:34:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6696eabd48bba547ef96402793a133651c8a6a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 03:35:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61126d14ae6ed9dab5e1704ce060506a1d04324e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:23:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a961348ec4224c8f99a8e1c8a3af3a59\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e90b9707b15b4d9690f7828792e0de24 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec98e38e87dc4ebd9b9d54b0a77281b8 : SUCCESS in 4m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c3cd5a0a629c4e49bc30366c7a163a2c : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a17f9a2e3bab40c2bcdc7a38893a369f : SUCCESS in 3m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9e8823c1e0ed4b94b72ce8b912f6aae4 : SUCCESS in 3m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ea0ba8f635174b71bd28cea52b478e7c : SUCCESS in 43m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a1b77ec0e8eb41deaa4722c8b9e66cea : SUCCESS in 29m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fedff688df564654acb76396bd78d8eb : FAILURE in 47m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/93a82a75de6f4827878bf92b8bee6349 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a3e0fbe67fedbbf3750e89c364df8dc856eebd6b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 09:22:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9032ec8cdb99927ecc0d66b4e340754cc5c45d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 10:42:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdab5ad5f5594b019435a19d270e2e7b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d016f37d308b4313920e875d3166f37b : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b996f377d34aad9bd5b7427bcad191 : SUCCESS in 8m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d51697bc1d094390b42b128172ff05fc : SUCCESS in 11m 25s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0e4bde4b7d0f4b72888843438cc7700c : SUCCESS in 7m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/338513b3c1fa4e69a5c3d1dfab965d78 : SUCCESS in 5m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a07873dbb202409bb32dd004ab131d3b : SUCCESS in 1h 12m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7d22942c449047a589e39a05ef534d48 : SUCCESS in 48m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f889a5adf0bc4a2ca66295f3321086ce : SUCCESS in 38m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b2720d22a7764b4aa8e623e67d811c92 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"47b1912710e003f5e37ac77dfb80cf5f350a6962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 10:43:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c73007949b60d9c42461c89f877403a00211723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 11:54:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/752d0cfeb1684d81a8c75606c56181c4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3accc7bf04f143429a70e7485feed3c2 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65cd4b3533c54506a46980b849eca102 : SUCCESS in 6m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a3af12f3a9ea4446b89997f258360547 : SUCCESS in 11m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ad77f6db76444409af1d36402f402674 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0de179e37cea470f81f76e22a7e43c9c : SUCCESS in 8m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6c0da575c28d48cb9cfef1aabf3cccc2 : SUCCESS in 1h 06m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b1f641004b7e4215b6d684db90daadc8 : SUCCESS in 1h 02m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1b3827007fa048fcbe156b7e3baca44e : FAILURE in 46m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/78883a562a32469b89b38dd13d290890 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8f32a3e153ada2f438b494416d37dec9cf8747d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"real_author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-26 09:30:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_38396\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"}],"_revision_number":2},{"id":"8dbbf70e5b08d114d0b9d9d55377478595e40b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 10:44:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc504c8e2c2746f6a67ece6953ac2dba\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd1a502a94a74ff69f3e96e878bf67fe : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c479960164ce4a93b9c84f1539c14e0c : SUCCESS in 7m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f5017a537e704c61ab5cae4e785b3dc6 : SUCCESS in 8m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a27dd6a044614b40ba1b08053c8ac064 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a0b97991bf9f43d3a6cc277f66769d52 : SUCCESS in 5m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1876ab58a86a486e80357b86a4a680c4 : SUCCESS in 1h 08m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/14df2d8926bc4318b1124a981a4b3a89 : SUCCESS in 42m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e01b028194674391805b119cca7f04f7 : SUCCESS in 49m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/25b853e18e0e4ac4a00920850712dd46 : SUCCESS in 49m 31s","accounts_in_message":[],"_revision_number":2},{"id":"340fb481cd698a9919e4d34001fa8cddac908e33","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-26 13:04:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a9f318dba998a8636a5f37b9d52c71449002cb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:05:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"183985cf599febbf73b1914188319890640457eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 14:21:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c77e8215c44314aaadf36df61c315e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42a3a02ae0ca413987f79e7eb63ed498 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fc46bb5dd14af0848b492c848a3a74 : SUCCESS in 3m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f0c2ade69bfe4d1e927cc54e7e1a2aa2 : SUCCESS in 9m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c7e3752fe81b45e0a0b571fde4870bc1 : SUCCESS in 7m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/57edfa8aa75b40e08814ce27c3822e16 : SUCCESS in 4m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/767c995ba9504bc987240564db4f002a : SUCCESS in 1h 10m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/32d9362523484eec81d38c0caf0c2372 : SUCCESS in 30m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a6551665c7a94bed9bcb4c4b13b376e4 : SUCCESS in 37m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6d2a66f97202449ab4bf529c4a840ed3 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"682a18250dfb67ff00ed1c8a3ca05c8413d72b2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 14:21:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"23603929031797e9588810a37e01273195025d97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 14:22:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73e333267ea489b817d9c4f26496319\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eedaf542130477e98d3c47f9c5677dd : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d3daefeaa7b08890f60c721b0ad1eda9f1cf96b","revisions":{"d644e011a547d9705b23e58b242e90d793baa726":{"kind":"REWORK","_number":1,"created":"2026-05-21 12:50:59.000000000","uploader":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"ref":"refs/changes/27/989527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/989527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/1"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"Rebrova Alena","email":"rebrova@ispras.ru","date":"2026-05-21 12:35:44.000000000","tz":180},"committer":{"name":"Rebrova Alena","email":"rebrova@ispras.ru","date":"2026-05-21 12:49:42.000000000","tz":180},"subject":"Fix magic input for only options-filters","message":"Fix magic input for only options-filters\n\nIgnore a string-input when there are no string-filters and open the filters dropdown.\nCloses-Bug: #2153560\nSigned-off-by: Rebrova Alena \u003cscahann@mail.ru\u003e\nChange-Id: I8b0e3c81590a646954c4997ec227c05dfc2bd5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d644e011a547d9705b23e58b242e90d793baa726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d644e011a547d9705b23e58b242e90d793baa726"}]},"branch":"refs/heads/master"},"0d3daefeaa7b08890f60c721b0ad1eda9f1cf96b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-26 09:30:03.000000000","uploader":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"real_uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/27/989527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/27/989527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/27/989527/2"}}},"commit":{"parents":[{"commit":"2e2fd6348744e1dc7e1c29ae00677fd4d6a5bc07","subject":"Merge \"fix: Empty Remote IP Prefix for remote group rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2e2fd6348744e1dc7e1c29ae00677fd4d6a5bc07"}]}],"author":{"name":"Rebrova Alena","email":"rebrova@ispras.ru","date":"2026-05-21 12:35:44.000000000","tz":180},"committer":{"name":"alena","email":"rebrova@ispras.ru","date":"2026-05-26 09:30:03.000000000","tz":0},"subject":"Fix magic input for only options-filters","message":"Fix magic input for only options-filters\n\nIgnore a string-input when there are no string-filters and open the filters dropdown.\nCloses-Bug: #2153560\nSigned-off-by: Rebrova Alena \u003cscahann@mail.ru\u003e\nChange-Id: I8b0e3c81590a646954c4997ec227c05dfc2bd5cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d3daefeaa7b08890f60c721b0ad1eda9f1cf96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d3daefeaa7b08890f60c721b0ad1eda9f1cf96b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d644e011a547d9705b23e58b242e90d793baa726","theirs":"2e2fd6348744e1dc7e1c29ae00677fd4d6a5bc07","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
