)]}'
{"id":"openstack%2Fskyline-console~989532","triplet_id":"openstack%2Fskyline-console~master~I67afbb95b2016ba3c39d17130e8745785db34b0c","project":"openstack/skyline-console","branch":"master","topic":"bug/#2153682","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-05-25 02:58:07.000000000","reason":"removed on reply"},"38396":{"account":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"last_update":"2026-05-25 04:08:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67afbb95b2016ba3c39d17130e8745785db34b0c","subject":"Fix frontend sorting in the registers","status":"MERGED","created":"2026-05-21 13:35:25.000000000","updated":"2026-05-25 04:09:32.000000000","submitted":"2026-05-25 04:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989532-bug/#2153682","meta_rev_id":"da76a6a53a2b344ba00f526a07c01595683d8bb0","_number":989532,"virtual_id_number":989532,"owner":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 04:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-05-25 02:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-21 22:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-05-25 02:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-21 14:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-21 22:12:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-21 22:12:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"29b3ecb4228021568806b2e0a352c374abff3727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"date":"2026-05-21 13:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17273dc6d8d7fdf0607b638ad7df33b27a57942e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 14:48:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1686d2b11a49828636ea8f585a72bc\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad24f25af520412baa2ec40d35644b84 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca558034b33416e962e27b24208fb01 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0e29ee4e5de14818903bf7c5c1af837d : SUCCESS in 6m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ee6f459926d442aeae291eb5b68374a3 : SUCCESS in 7m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c928c841cd3c4b8eb3f2bb4e402214b3 : SUCCESS in 9m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ed2f9a5c5dd34ed1a2a724fe0c8bf72c : SUCCESS in 1h 07m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/473a60cef99a4e0cbe97ad2cd3030dfc : SUCCESS in 45m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cb413afdc54548489b478e4aad529f26 : SUCCESS in 58m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e05ce98397f04efb8146062c05076afd : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1e4eedfa9fc4708f46a12d780786246b6e3f0f68","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-21 22:12:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14c36847fd5d66198bc2af2abb4576f8b3d13a09","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 02:58:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75712c982e83653f5914f182b9c3d352010aa73c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 02:58:14.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0078f0e0e61dcae126f84e75a5dd43c4f5b5d487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 02:58:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"310f991c7c624cb4e1c71d3f3740afa98a3c0dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:08:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8589cf5e034449f974d3e0324fa618d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9459639becc04a36b2ec5fb60e32e6e6 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4326870d16948ebb72a3d0076d02fb2 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5b053f94e4a7498d8da495720d6a3177 : SUCCESS in 8m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bd04f8dbd3954d4f84a6634602de9bcf : SUCCESS in 6m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/aa070b303a844235b7ee54e23ba78a24 : SUCCESS in 7m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/266d43332e044f2195a90aaeb9b49f00 : SUCCESS in 1h 09m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ea932cf87d2a4ba481e463a3e9d1cd54 : SUCCESS in 55m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d619be74cd914c59b4d6e78027f142b8 : SUCCESS in 54m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/11e3f517db834e4e9f9829f27f06c705 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f1b02909d49c3ebd05f23c6bebfe495dab86141a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da76a6a53a2b344ba00f526a07c01595683d8bb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202a44e80e38481d9dbdf06e506d7a17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e1ac96280b450bbbc96b793437a576 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b91efc16dac016273bcd23b6c444f9f81c0980d2","revisions":{"b91efc16dac016273bcd23b6c444f9f81c0980d2":{"kind":"REWORK","_number":1,"created":"2026-05-21 13:35:25.000000000","uploader":{"_account_id":38396,"name":"alena","display_name":"Alena Rebrova","email":"scahann@mail.ru","username":"scahann"},"ref":"refs/changes/32/989532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/32/989532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/989532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/989532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/32/989532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/32/989532/1"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"Rebrova Alena","email":"rebrova@ispras.ru","date":"2026-05-21 13:34:17.000000000","tz":180},"committer":{"name":"Rebrova Alena","email":"rebrova@ispras.ru","date":"2026-05-21 13:34:30.000000000","tz":180},"subject":"Fix frontend sorting in the registers","message":"Fix frontend sorting in the registers\n\nChange-Id: I67afbb95b2016ba3c39d17130e8745785db34b0c\nSigned-off-by: Rebrova Alena \u003cscahann@mail.ru\u003e\nCloses-Bug: #2153682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b91efc16dac016273bcd23b6c444f9f81c0980d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b91efc16dac016273bcd23b6c444f9f81c0980d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
