)]}'
{"id":"openstack%2Fskyline-console~989918","triplet_id":"openstack%2Fskyline-console~master~Id5f27fc88680699c00d89bf10f99b7105c74d970","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"Id5f27fc88680699c00d89bf10f99b7105c74d970","subject":"fix: Adjust quick search label for instance port filtering","status":"MERGED","created":"2026-05-25 02:52:18.000000000","updated":"2026-05-25 05:09:31.000000000","submitted":"2026-05-25 05:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989918","meta_rev_id":"b015ffaa74bde09b41a03d306ecf376bf0c1fb56","_number":989918,"virtual_id_number":989918,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 05:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-05-25 04:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-05-25 04:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 04:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db78f75a635793054ca4ac911bcbb0e89ebf071a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 02:52:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3124e4be906253150259e3c0c5e27c37213204a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 03:03:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37f5438ea70a87cb5dd276b2243258a2e3814b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:01:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2523da35dc054449bb97cc197ee1ba88\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7487c530b8d4705b543a1a0881dcbb8 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723c26262e2e4a9ca3ef2197f019babc : SUCCESS in 6m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8cbadef4be834e13a425a5790f1b0139 : SUCCESS in 10m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/20baaf9d5f92440aa8f8b62580eb1c07 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1056ec77a13c43169917561e7f4205ed : SUCCESS in 4m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f2c2a474475046ee8890570156fac273 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"87bbc69e88ac497587e06b2a534d7eb8187359b4","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 04:04:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"325cdfe900dffd27c9725094efea473b995d4036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:05:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9fcbbc1a580e9ce30397ab99e0e687ea65500c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 05:08:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b497fec036447a9d768d6f663ed6af\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a287dec9d3bb4beb98bb3aa6ede398fa : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e781c2fd3fe40939222810d62035efa : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b33dcef6e7cb40d4ad65549b486de5cc : SUCCESS in 9m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/603a8e1cb6834c70b8efd82475f61cf9 : SUCCESS in 4m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ad8246b5b3d04981a9b9b44995f43647 : SUCCESS in 8m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf856f4881464fedb0af09250ea66386 : SUCCESS in 44m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5a39eb6de49204f10f8849a141e584d392ab56a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 05:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b015ffaa74bde09b41a03d306ecf376bf0c1fb56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 05:09:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1852cb5837d74a7f97da592c545d09d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3defb9b3404b3997f9fee0864f7663 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0641af9a3e4b8277ed5159f45d7db0db634062c","revisions":{"b5224ebe414f5e7e7ae46a8576ec12d995c21a05":{"kind":"REWORK","_number":1,"created":"2026-05-25 02:52:18.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/18/989918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/989918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/1"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 02:47:55.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 02:49:15.000000000","tz":480},"subject":"fix: Adjust quick search label for instance port filtering","message":"fix: Adjust quick search label for instance port filtering\n\nUpdate display text of quick search entry for filtered instance ports\n\nChange-Id: Id5f27fc88680699c00d89bf10f99b7105c74d970\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b5224ebe414f5e7e7ae46a8576ec12d995c21a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b5224ebe414f5e7e7ae46a8576ec12d995c21a05"}]},"branch":"refs/heads/master"},"c0641af9a3e4b8277ed5159f45d7db0db634062c":{"kind":"REWORK","_number":2,"created":"2026-05-25 03:03:16.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/18/989918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/989918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/989918/2"}}},"commit":{"parents":[{"commit":"4ecddc901ae9191a75fe7bc701fe46d4291ee8c2","subject":"test: Disable ironic deployment in DevStack E2E tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4ecddc901ae9191a75fe7bc701fe46d4291ee8c2"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 02:47:55.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 03:02:25.000000000","tz":480},"subject":"fix: Adjust quick search label for instance port filtering","message":"fix: Adjust quick search label for instance port filtering\n\nUpdate display text of quick search entry for filtered instance ports\n\nChange-Id: Id5f27fc88680699c00d89bf10f99b7105c74d970\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c0641af9a3e4b8277ed5159f45d7db0db634062c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c0641af9a3e4b8277ed5159f45d7db0db634062c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
