)]}'
{"id":"openstack%2Fskyline-console~989929","triplet_id":"openstack%2Fskyline-console~master~If034fa5afcfd2016057275f1b6d67a4e4b68824b","project":"openstack/skyline-console","branch":"master","hashtags":[],"change_id":"If034fa5afcfd2016057275f1b6d67a4e4b68824b","subject":"fix: Show red overflow state in quota chart for instance/volume creation","status":"MERGED","created":"2026-05-25 06:46:34.000000000","updated":"2026-05-25 09:36:24.000000000","submitted":"2026-05-25 09:34:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989929","meta_rev_id":"ac69674e93e3cd5ea96657d73bac0ca3e9f7ed3d","_number":989929,"virtual_id_number":989929,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-25 09:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-05-25 08:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-05-25 08:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 08:05:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec6994785ab0c0f8ac8027b55a1c1d6cf504cd15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 06:46:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"544162618fee6dd5b0dde073a20566988519dafe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 08:05:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ea844039f54c1bbc8d04796b92abc7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/920be3e029f74f12bc2a3dcba9d31aeb : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa00e3384b614608a0e812097faa892c : SUCCESS in 6m 59s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/940919075c6b4806b97851d160161a02 : SUCCESS in 5m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bb1159a490d94d6aa6ae898922308206 : SUCCESS in 3m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/930a8c675fbb4128a782f7a2a50ae0f1 : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c36d3c36f052466584d345f6176e1db3 : SUCCESS in 1h 13m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e5f24a0604254b2ca523aef63bbe0788 : SUCCESS in 1h 00m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3737171b9964493b907c7d40e9dba901 : SUCCESS in 56m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2b46333972254798b123fa3204ce7251 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5e5ffde97902b9e44d03bec452876e6467126975","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-25 08:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"468be07824dabf4821ea6b037bfe445422cd70bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 08:20:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8410706f72ae5d769c8afbbb721f0c75205d960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:34:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e8ee3b4b0f4d2391140721921c03af\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/648a76e0b74d42e08de50771ee50bfda : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad2f1cd309c4406b75b0f1a353fcb6f : SUCCESS in 6m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3c8a63a2d8fe43849c9add718d65e985 : SUCCESS in 10m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f5a84fc3603f44139e7a6ac0b7d49ac7 : SUCCESS in 7m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8c196ba032e34a63bd5820989e0827be : SUCCESS in 3m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ddef6c6039d84b0e85cd34de6cfe775b : SUCCESS in 54m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/31c47194b52940e4b81f81dd08fe45ed : SUCCESS in 1h 02m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/288df9a9787f46e4adc117bf7f758aa9 : SUCCESS in 50m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ba48c6dc167f47918fb6c8b9dd1fbce5 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ef0fa54b32161e1cbab7b75dfae16ce3bcc3be81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:34:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac69674e93e3cd5ea96657d73bac0ca3e9f7ed3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b012afdf8d4109a5b8367abdb32e7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdbd336e5e04df0bc38faa310c83366 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f7723653e8af838b2a2cb258061fac9de595a56","revisions":{"3f7723653e8af838b2a2cb258061fac9de595a56":{"kind":"REWORK","_number":1,"created":"2026-05-25 06:46:34.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/29/989929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/989929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/989929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/989929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/989929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/989929/1"}}},"commit":{"parents":[{"commit":"ad7fd9f2514be41fb43ab2d85f33f234ce22dc9b","subject":"fix: Unify attach volume operations for instance list, detail page and architecture view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ad7fd9f2514be41fb43ab2d85f33f234ce22dc9b"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 06:43:37.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 06:44:34.000000000","tz":480},"subject":"fix: Show red overflow state in quota chart for instance/volume creation","message":"fix: Show red overflow state in quota chart for instance/volume creation\n\nFixed the quota visualization issue in instance and volume creation forms. When the requested resource exceeds the available quota, the quota chart will now display a red overflow indicator to clearly show the over-limit status, instead of showing incorrect or unchanged visuals.\n\nCloses-Bug: #2146904\nChange-Id: If034fa5afcfd2016057275f1b6d67a4e4b68824b\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f7723653e8af838b2a2cb258061fac9de595a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f7723653e8af838b2a2cb258061fac9de595a56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
