)]}'
{"id":"openstack%2Fskyline-console~990046","triplet_id":"openstack%2Fskyline-console~master~Ic85013ebab2c0ba59b66e6accab32d998451b9c8","project":"openstack/skyline-console","branch":"master","topic":"fix-e2e","hashtags":[],"change_id":"Ic85013ebab2c0ba59b66e6accab32d998451b9c8","subject":"test: refactor and simplify user E2E cases","status":"MERGED","created":"2026-05-26 05:55:00.000000000","updated":"2026-05-26 08:43:34.000000000","submitted":"2026-05-26 08:42:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990046-fix-e2e","meta_rev_id":"f6f6330cd29f59abcc9577af775c1950c33a8894","_number":990046,"virtual_id_number":990046,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-26 08:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-05-26 06:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-05-26 06:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 06:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26562ca40c169a7a879a454bce33fe74d2106dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-26 05:55:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"817332b28afb9345304d6851071b73b940851166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 06:58:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cc1f52878a8441e9c63bd3275f20b51\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a92a23d138c04d6f9d2317e0a7fd35f8 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d20005054a94b5095e98fcfd668e02e : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7b548624fa114cb891fb6ddcc7b7bba9 : SUCCESS in 6m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e4764c703c804c7babc06565bc22512c : SUCCESS in 7m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/882e6a6e3a9940e2a56cbbf2f67716b2 : SUCCESS in 3m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3bc716a694584f32a6e4d72590ddd071 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c33e89980601f02b4e818117da97deda93c66198","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-05-26 06:59:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba511cf3c0a056fec5f371fe0e688e14fdf4bbb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 06:59:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42e52ddeac080ca5bc3b22de8b1efd0d5e6544ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:42:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a96fd58359c345d086a283402d5b7a7e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c3655d89e374d4489b56a9d31642682 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32e3fade53d49dda296e2d854401097 : SUCCESS in 6m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cc4e6e00c67840b7bf8c04c4370d6578 : SUCCESS in 8m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/969bc4a7a50f468aa6f20daa34709c96 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3671640c761c44b1b97c9b684e754d02 : SUCCESS in 8m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3a7f582699cc4b0fa2474df2414919c4 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2a3baa4fb6fe2750cca041c8f1e4163ab8f754c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:42:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6f6330cd29f59abcc9577af775c1950c33a8894","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:43:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97f245de2e8464fbd3f81eca6c3db54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6566d0b499f84b218541378accaab08f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6d7234c94a2975130babaaff1bf8d86d413c6a6","revisions":{"f6d7234c94a2975130babaaff1bf8d86d413c6a6":{"kind":"REWORK","_number":1,"created":"2026-05-26 05:55:00.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/46/990046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/46/990046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/46/990046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/46/990046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/46/990046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/46/990046/1"}}},"commit":{"parents":[{"commit":"3f7723653e8af838b2a2cb258061fac9de595a56","subject":"fix: Show red overflow state in quota chart for instance/volume creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3f7723653e8af838b2a2cb258061fac9de595a56"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-26 05:52:19.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-26 05:52:56.000000000","tz":480},"subject":"test: refactor and simplify user E2E cases","message":"test: refactor and simplify user E2E cases\n\nSimplify user E2E test cases to only keep basic CRUD operations.\nRemove irrelevant project and group association operations to make\ntest cases cleaner and more focused.\n\nChange-Id: Ic85013ebab2c0ba59b66e6accab32d998451b9c8\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f6d7234c94a2975130babaaff1bf8d86d413c6a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f6d7234c94a2975130babaaff1bf8d86d413c6a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
