)]}'
{"id":"openstack%2Fskyline-console~990448","triplet_id":"openstack%2Fskyline-console~master~Iaf1ec80051bf9f8122867d2e22fa22dd8564da32","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"38366":{"account":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"last_update":"2026-06-04 12:11:47.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-04 11:28:10.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iaf1ec80051bf9f8122867d2e22fa22dd8564da32","subject":"Fix hypervisor display for dedicated CPUs (pCPU)","status":"MERGED","created":"2026-05-28 11:45:29.000000000","updated":"2026-06-04 12:12:55.000000000","submitted":"2026-06-04 12:11:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990448","meta_rev_id":"07fc17ab8235cb2adf0d2b41d37fdbe4e035cf48","_number":990448,"virtual_id_number":990448,"owner":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 12:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-06-04 11:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-04 11:28:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-06-04 11:24:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 13:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-30 16:25:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-04 11:23:54.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-06-04 11:28:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"d01df8cea68c015228175aa9cfee81ac8f15d152","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-05-28 11:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"042f1bb65662b7715c81ec56c079d24c560ea3f9","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-05-28 11:47:26.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"1f392551a7c14b04044896e54f33f6a1651c6b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d732ab224b04baaa4aecb3c4f500bfd\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/356ae3d53b124e03a1e3fe298ff39e7f : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e90468f45a452c957f37e735a69121 : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6cae4b5b9c4b4da0843b5ca90dccb6f2 : SUCCESS in 12m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53c7fdb842274aea84d8822563c1185c : SUCCESS in 4m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/379a89a67af54123b0bf48fabb89aea6 : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d7dcdd50fbc34e9f9083de52d3bbe22a : SUCCESS in 1h 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"53d31f029ab6448a894686940d2472c942a2130b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 16:25:58.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb175a8f41a711d2a399294ab0ce1cc00d64ed97","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 16:36:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9d9d89ff8b6143108bda6db9c3b2a3ff4363994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-06-01 13:53:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"730778cb44b2651c03a7e4aad9f927e889fe3386","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-06-01 13:54:34.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3f079cdf8e5222e2b56a882f320a64253b08f2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:40:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e342d94565284a3fa0d9a3efd4efff92\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3facdbc6b80c432c925a75a1fc06c450 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2474f773c7a24781b2d47e853518c454 : SUCCESS in 7m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/70febc7e56e047668fd9e6cb85ba4741 : SUCCESS in 4m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8225c8367a6240759a5b9bb9671ff94f : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/92d4d0d16fa74bc88d8dc3c4c7284b9d : SUCCESS in 6m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/84ab8036356f4d1581c04432a7d38629 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3f317e9cb1957e23060dfd10f5dbadfce9242d3e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-06-04 11:23:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21374dc10e00ff88014f101ae5e290ec10e0ea14","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-06-04 11:24:01.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5b4a62df1c3f47e34df6c141e13690ac348e6ee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:24:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f91ad4e06bb5f787d55b00c47c6ac172ab9f27b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-04 11:28:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cd789dd0f97a2904f2e450cecbb8ca5820e808b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:11:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7305d820a7841b9809f03ec5939ee7e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8110a90299504cb38cccd1d64a323435 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d078035024f04d73b417b4f319e131d9 : SUCCESS in 4m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74319a4cfb614631a2379b5c1a9edc50 : SUCCESS in 4m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/222d8a674ac84516ada9e9809624db84 : SUCCESS in 5m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7aa1f3dbacef45968c522f4152f8be9c : SUCCESS in 4m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c16d6c642c38441bb3904397678c6ae4 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f8e17fdd255b0b09f72961dfd05c63844dbcaab3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:11:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"07fc17ab8235cb2adf0d2b41d37fdbe4e035cf48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:12:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55386865652f421ba0d99fa4aeac1a2e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb322c964bcd4da3b6dc1556e3575965 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e750917c744a4b81c8e4564fde4cb0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ee76d7a2050523820481a3a6f904f36d0288cb8","revisions":{"3e848f7a772a8a8e1155b59317417312eeea5e57":{"kind":"REWORK","_number":1,"created":"2026-05-28 11:45:29.000000000","uploader":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"ref":"refs/changes/48/990448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/990448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/1"}}},"commit":{"parents":[{"commit":"ec33c17046a291e0a8d94de00a0df221e5e78ad2","subject":"Merge \"Set browser title from current page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec33c17046a291e0a8d94de00a0df221e5e78ad2"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-05-28 11:36:07.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-05-28 11:43:50.000000000","tz":120},"subject":"Fix hypervisor display for dedicated CPUs (pCPU)","message":"Fix hypervisor display for dedicated CPUs (pCPU)\n\nFix crash when viewing compute hosts with dedicated (pinned) CPUs.\nThe code assumed VCPU inventory always exists, but dedicated CPU\nhosts sometimes only have PCPU inventory in Placement API.\n\nChanges:\n- Add null-safety for VCPU/MEMORY_MB inventory access\n- Add support for PCPU inventory (dedicated CPUs)\n- Support mixed configurations (both VCPU and PCPU)\n- Add cpu_type property (\u0027vCPU\u0027, \u0027pCPU\u0027, \u0027pCPU+vCPU\u0027)\n- Update UI to display CPU type in list and detail views\n\nCloses-Bug: #2154443\nRelated-Bug: #2107470\nChange-Id: Iaf1ec80051bf9f8122867d2e22fa22dd8564da32\nSigned-off-by: Maximilian Sesterhenn \u003cmaximilian.sesterhenn@wiit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e848f7a772a8a8e1155b59317417312eeea5e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e848f7a772a8a8e1155b59317417312eeea5e57"}]},"branch":"refs/heads/master"},"7ee76d7a2050523820481a3a6f904f36d0288cb8":{"kind":"REWORK","_number":2,"created":"2026-06-01 13:53:36.000000000","uploader":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"ref":"refs/changes/48/990448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/990448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/990448/2"}}},"commit":{"parents":[{"commit":"ec33c17046a291e0a8d94de00a0df221e5e78ad2","subject":"Merge \"Set browser title from current page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec33c17046a291e0a8d94de00a0df221e5e78ad2"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-05-28 11:36:07.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-06-01 13:52:44.000000000","tz":120},"subject":"Fix hypervisor display for dedicated CPUs (pCPU)","message":"Fix hypervisor display for dedicated CPUs (pCPU)\n\nFix crash when viewing compute hosts with dedicated (pinned) CPUs.\nThe code assumed VCPU inventory always exists, but dedicated CPU\nhosts sometimes only have PCPU inventory in Placement API.\n\nChanges:\n- Add null-safety for VCPU/MEMORY_MB inventory access\n- Add support for PCPU inventory (dedicated CPUs)\n- Support mixed configurations (both VCPU and PCPU)\n- Add cpu_type property (\u0027vCPU\u0027, \u0027pCPU\u0027, \u0027pCPU+vCPU\u0027)\n- Update UI to display CPU type in list and detail views\n\nCloses-Bug: #2154443\nRelated-Bug: #2107470\nChange-Id: Iaf1ec80051bf9f8122867d2e22fa22dd8564da32\nSigned-off-by: Maximilian Sesterhenn \u003cmaximilian.sesterhenn@wiit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee76d7a2050523820481a3a6f904f36d0288cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ee76d7a2050523820481a3a6f904f36d0288cb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
