)]}'
{"id":"openstack%2Fskyline-console~995030","triplet_id":"openstack%2Fskyline-console~master~I4cb8adb3f9539eda0728a538cc7d49740746db17","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2026-07-07 05:19:31.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:21:08.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2026-07-07 05:19:31.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2026-07-07 05:19:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4cb8adb3f9539eda0728a538cc7d49740746db17","subject":"fix: pass username and domain separately for TOTP setup","status":"MERGED","created":"2026-06-26 11:35:17.000000000","updated":"2026-07-07 05:21:14.000000000","submitted":"2026-07-07 05:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995030","meta_rev_id":"8a5128af9bc27b49db6aac1d6d036d0d87b3679a","_number":995030,"virtual_id_number":995030,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 05:19:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-07 03:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-06 15:03:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-07 03:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"CC":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 12:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-29 06:47:57.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2026-06-29 06:47:57.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-06-29 06:48:32.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2026-07-06 05:46:20.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2026-07-06 15:03:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"be228cff51034cc538fb1849a7117efa5a8ff3b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-06-26 11:35:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b923753e1e5337d75dc6abac29c3ec1c46c71058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 12:30:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20779cb89f064058a1b01321ce719066\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a53641f72c954741a34b5dc04f9e49cd : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0947a1db354f459daf272eff93c6b179 : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c90d777e8e4b41eca389650b4127f1ae : SUCCESS in 11m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/39029524d432478a852458187f0e092b : SUCCESS in 3m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5d41b0c1742d4014b1a861380fc0ac1c : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5ecc6853b8a6426faf9a0c20cb345f84 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":1},{"id":"88693a650d109fd2229f128c10328eb53ed057b2","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-07-06 05:46:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77a4efd87beec3d087b54aa35d5c57419783a22c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:03:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c1af8d87124806691f87ce8eed4877fe304317d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70054263b6c24762fe12647afea2fd6db96ea232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6acf1581dd84f547ac9a9a52b28f6731c6e851a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:19:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fc4eba55a4465e872939d453c3db58\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9ae3ecb74ee45b69dbcc65ab1d0d194 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a07a35ed8943a3ba8abca948056d30 : SUCCESS in 7m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5506749eeb874539a67038d6b946a717 : SUCCESS in 12m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f604fbdebc6f431f86a8c2ae4e45a68a : SUCCESS in 4m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/09012b32749341269165d534c142da2c : SUCCESS in 4m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/beea8cbb2a3b49a283859a92890f1e35 : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e9b6c538dde30f11400a2549ec0a66bbbf26d1dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:19:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a5128af9bc27b49db6aac1d6d036d0d87b3679a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5965dc17c3064d63b390bcea4c5227d0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f4bd879ad24ff799f5b2044c17378c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe517d891d75e8e3857b8a97f7db00e0a368a035","revisions":{"fe517d891d75e8e3857b8a97f7db00e0a368a035":{"kind":"REWORK","_number":1,"created":"2026-06-26 11:35:17.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/30/995030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/30/995030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/995030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/995030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/30/995030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/30/995030/1"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-06-26 11:31:27.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-06-26 11:35:24.000000000","tz":0},"subject":"fix: pass username and domain separately for TOTP setup","message":"fix: pass username and domain separately for TOTP setup\n\nChange-Id: I4cb8adb3f9539eda0728a538cc7d49740746db17\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fe517d891d75e8e3857b8a97f7db00e0a368a035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fe517d891d75e8e3857b8a97f7db00e0a368a035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
