)]}'
{"id":"openstack%2Fskyline-console~995825","triplet_id":"openstack%2Fskyline-console~master~I3a4006350437394235efef6a179a2a634871c595","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:14:48.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"38366":{"account":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"last_update":"2026-07-07 04:17:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a4006350437394235efef6a179a2a634871c595","subject":"Add clouds.yaml download support","status":"MERGED","created":"2026-07-02 13:49:57.000000000","updated":"2026-07-07 04:19:09.000000000","submitted":"2026-07-07 04:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995825","meta_rev_id":"f676bdd71163cc919b9ab48bb33edf3a6ce9264e","_number":995825,"virtual_id_number":995825,"owner":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 04:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-07 03:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-03 01:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-07 03:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-02 15:13:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-03 01:17:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-03 01:17:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b05d05e2a54d2a49deb2e768091d4f34b5c2056b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-07-02 13:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb31cad2114574425767bd2f5cecb4e7dc2354ae","author":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"date":"2026-07-02 14:00:39.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"959801c23b376166800d8cc0a417dc721ca18b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 15:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc01fc7032f47edaef78f9bb32b61fb\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cc3ab6ba2924dfea9bf6b3629d9df6d : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b74d2f87fa4bf7b30e3e7551bef00e : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/24b27ad645be481686563bade46dadae : SUCCESS in 10m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0a8d979942bc4704951c244a25240b58 : SUCCESS in 8m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e07e40b57bd54a3188ffb6a1faf3cfad : SUCCESS in 4m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6b47eaaafbe14baf8374dd5ecf297126 : SUCCESS in 1h 17m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7205985fe0da47078e8f6f44abbe743f : SUCCESS in 1h 05m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/db3d2269b1284df48e860486edc083f3 : SUCCESS in 51m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ce52c9e58cb746dfb35b6e6c8bc06361 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8b6f13a902f12ab689302ad73104bbf3596076ac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-03 01:17:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"af03e48b7326bc951999072769b34c91596f2a51","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:14:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cfc965b86acb3a882540b11c76c49a707bc683b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:15:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22132183aecfb3127bb0775508913efa878be46f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:17:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c36dfe5c6b244afe8396d088651de703\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0330a82af84047e29777f472462e4249 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9998c32203414444a2a5e47b56db81f7 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/23aa27a4f24445ad837c5af00e68da1a : SUCCESS in 11m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/57511f42caf348da96b0b189e1ccfdd9 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bac4e0aa34e8411bbd6e1282509359ab : SUCCESS in 4m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7a05b95787bf4484ad75ee7ef48156e4 : SUCCESS in 40m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a955425c6dc94f6d842bd23f41c2e47a : SUCCESS in 47m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a603da9f5bb0425c87683b513f76309a : SUCCESS in 42m 56s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/efb84dfaf4cb4713a5e6d439b8f527ed : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2b77ba0c02eaaf5dbf8ae7b94da310f770488bf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:17:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f676bdd71163cc919b9ab48bb33edf3a6ce9264e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:19:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8e306453cd46f1a9303cd884d16727\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41a7e86691014787b3a7227ecc150d40 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/842539805e004160b9afe43c8e21a948 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d7c11f1d989387797a9bc8eb88e2cdfa3c15645","revisions":{"7d7c11f1d989387797a9bc8eb88e2cdfa3c15645":{"kind":"REWORK","_number":1,"created":"2026-07-02 13:49:57.000000000","uploader":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"ref":"refs/changes/25/995825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/995825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/995825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/995825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/995825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/995825/1"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-07-02 11:06:20.000000000","tz":120},"committer":{"name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","date":"2026-07-02 13:24:26.000000000","tz":120},"subject":"Add clouds.yaml download support","message":"Add clouds.yaml download support\n\nAdd a \u0027Get clouds.yaml file\u0027 option to the user dropdown menu,\noffering both password and application credential authentication\ntypes, matching the existing OpenRC download pattern.\n\nCloses-Bug: #2158966\nChange-Id: I3a4006350437394235efef6a179a2a634871c595\nSigned-off-by: Maximilian Sesterhenn \u003cmaximilian.sesterhenn@wiit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7d7c11f1d989387797a9bc8eb88e2cdfa3c15645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7d7c11f1d989387797a9bc8eb88e2cdfa3c15645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
