)]}'
{"id":"openstack%2Fskyline-console~996020","triplet_id":"openstack%2Fskyline-console~master~I873ccd29c1c833eff3514531cf0908f7e51c2226","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-06 18:47:23.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:22:38.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-07 05:22:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I873ccd29c1c833eff3514531cf0908f7e51c2226","subject":"Update Russian language","status":"MERGED","created":"2026-07-04 19:19:19.000000000","updated":"2026-07-07 05:24:11.000000000","submitted":"2026-07-07 05:22:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996020","meta_rev_id":"6d317b4fccd9d5a1bf5005b576466aa13612ed5d","_number":996020,"virtual_id_number":996020,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 05:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-07-06 18:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-07-07 03:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-07-07 03:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-04 19:32:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-07-04 19:32:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-07-04 19:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-06 14:57:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"1e81fb77aff02782cdd00e6aa96c6ddbfa5f71f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:19:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b74edc02683e28d6ace9f83c1fcfacf7918f11e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:31:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ac7516c61c1ec5f8c6712cf2b02d74416c139251","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:32:13.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"828a90a5f643012ee8ac2cbe8ebab4fd22448248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:32:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a53352c51a60d194d08b1c01867a9fc6188c7c5d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:32:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"d189193ccac7bfb3c773abe1d48eb808b65bb4cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:34:34.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"06ba5fecc9e288c62bdcebb7df75ab41104cb991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 19:50:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a43cd8c5dd48a0badbbbf4ed898182\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4cbaa4b9f84c39921af5e00fe7677b : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952af797646a44b2876fabb979b8ba63 : SUCCESS in 6m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/211761d912d7464bab49e3f24df39d8c : SUCCESS in 10m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b0c15615e9dd454783314fb49eb7648a : SUCCESS in 4m 30s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b2832ae5f7d14232b1d693a8c8739ddf : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":4},{"id":"a2c3bfd3cf2a5fc8e4a3e96cf42078505c79af93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 17:15:59.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f7762f227309191fbd946dde651ea393a9824b22","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 17:22:52.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"728766f478265cc843ddbcca5f25ac0037c64f57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 17:33:39.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"66969711de6c335ec2006ac9aae5b90664386734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 17:42:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3a2ccf5c844e67bd4ae9da2713b13e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dfd5ac1a86a4eaf968b6a7556e59b0d : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e9ff6aa4f8457e971a7bb690db5659 : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/77b2104cccc847779e4eb9af6c8eaa58 : SUCCESS in 4m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/550be3ea2c184ae0bba93a9ef67014b6 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/331754cb120644198479e6fd9ce533fa : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":7},{"id":"aa13955d7dfe3579c3b0dd73385d90e2f246d8d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 17:55:26.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"24fe3fa933b014772f6cdea2333a0fe8e692b421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 18:08:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94303197650645e2b57679d09ed22b60\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/892abffd8f47427a817f8d683dd5f74f : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f212099d525f4a27abcf3473f6e09ff5 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3d8851badb464af28e8ce7bdb8e05149 : SUCCESS in 6m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/69e9b237f60649ff909f8e5124475d76 : SUCCESS in 7m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/05c6b3c083064ba9978cb26f5799a68e : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":8},{"id":"6432936e804263cc1a2e5d7a7d9cf4101c4a1678","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 18:43:43.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"8cc79bfcbbaf2af021a80625e8ff472439fcf487","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 18:51:46.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"a49563bd9924a03b4f72fd5c3a3f035832a4452f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 18:53:57.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"eed1b40ee7c8dd4a0fcfa2c81d808a717108ed6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 18:56:16.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"ab64da892cf2311544a56f418e8b8ef0e4daebc9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 19:03:59.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"462cef2e60c41b78747071d50f2ba9b90d69e1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 19:20:15.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/291e19484f90402596ed9409c070d995\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec439ff818634a9582319fc665e15449 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2eca8d95db84bfe8a990cb297f7c0cf : SUCCESS in 3m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/55ea1244d0a94b88b7491e19acdb182f : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d8c3b77c4fe44d04817949b419ce2030 : SUCCESS in 7m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/09fe5b092d48423997051e8083155929 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":13},{"id":"2f70c9b5fe4c3da6a3453b6115c6c6a9f02fd7b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-05 20:39:01.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"338218a915e739f9422c84dd3ddf3b93999956b5","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-05 20:39:12.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5b68119188f4833d2d0228158dd21bae767b77b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-05 20:46:45.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"2670c5bd61cab3971c31acb69d8bdf3ce4b2e518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 21:03:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f7a9689d3f4f3fb4f6389ba844cc5c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47711cf0f0794a5fb52cabf26624bdba : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3894b1eac45949a3b200c41a94da39d1 : SUCCESS in 7m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bf8f53dbfabe43eea9487fcfbc130abf : SUCCESS in 12m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/046bde9e3fd7462e919ed03d78f80f43 : SUCCESS in 3m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d3cca2788d764667be426046cf6c0e17 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":15},{"id":"5ab915d2762eeb35dbdace40c19560e093bbb1b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 21:10:27.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"d9cf8ee8478ffe08c96245b167344fb290cba3ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-05 21:18:15.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"e140514a7fc7026a93e06d0bb0b94882c8ce3d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 21:31:19.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803e05d4df33410cb8a3f935cf863a1d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a99d5f310b4a49d3abd6ed01700cb74e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766927879524439f87cb440d8da3b0e7 : SUCCESS in 3m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a5a8c668f4004e0599f82c6bf194a830 : SUCCESS in 4m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/236db9ee9d7346b7be880ba3ac719877 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/86c1149d1da64e2383c76fc6c0f3014b : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":17},{"id":"cb4e538ccff687a7e545d80063b0ce0396cb94fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 21:48:41.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"55f42b5e0ca469b6de292d1ffc6894b5d7ef1c16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-05 21:50:15.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"c28e94b6f52c4a3e7ced96f3ec7e3ff3d1d4de36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 22:02:14.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db558b948c845c5aae53f1a70679d16\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c1638d1dea42bab77bcd2fe293e3fb : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcddc3364944b86922602acd12ca47e : SUCCESS in 5m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d05ed4c7e5d0419ba1c97a049ebb856d : SUCCESS in 7m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/737b307a807642748e227adb88817a96 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/99af129183b446c5959cfa9511293e6e : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":19},{"id":"8bc4b6bc64afa567a7bb59a94f7f33eb48872300","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 11:43:07.000000000","message":"Patch Set 20: Published edit on patch set 19.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"51ea081e7677e30e6c677695cb2a9d329ddc0698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 11:48:37.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070c214b65d84fc492c69f873519a376\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e0ebcfd5f44048affa24ec7267b0db : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecfeeba29ae4c33b6f508d7b3dce788 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c5240722b78b4a20a8935aedf7108c4d : SUCCESS in 4m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/702ced7bcd73423994fd0d7d0c47037d : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/79ff3e05373742e1829a5619fdaf4f77 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":20},{"id":"4a994f005e0cb267ae5c9ecc3a4710b6bf1f1779","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 14:53:56.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"2808f39d7296686aa86ea242879a063d94969e2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 14:55:05.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"b4fe605213df799fbd4d779b7989845dd9f688c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 14:56:56.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"a9d24ac58321fcf1e84ed74ca890ba1e6c72063d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 14:57:03.000000000","message":"Patch Set 23:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":23},{"id":"98075017758b9f1fa2390f62235ba7eefb8e0a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 15:13:48.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435a101ea61d4e5eb9124bfd6326b4fa\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef069946fae449f5add0458ec9935714 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520c5ce47dfe4480b760625b9bd89fa5 : SUCCESS in 5m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e460c5961ac948468d084d09a9fe8be5 : SUCCESS in 11m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4b5feb950e804135a6401e9b820b0001 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/76bd8e9c9e4042b3bb9d1096d6937bba : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":23},{"id":"a720eb45cc5f34c4531a1226e172e0fc2a86f687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:20:34.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"58a56d95e6e23da27244c2ae6a686bc0ea5a7f2c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:20:58.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"c0e59cb913de56885e05cdb23d0092bb90374790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 15:35:20.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a8cde3d45d4f4bb2e6572cdbfa63ed\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/529cda8a071944d2a9637f8807f9e64b : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2fc3b518ae437aa513abb7fb34b023 : SUCCESS in 6m 39s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4464730eb995425db56132679ded046a : SUCCESS in 8m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/52e2300310fe4f5baf2ba809929f4894 : SUCCESS in 9m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f2ccef3bf9d54c279935deb4f4a6239d : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":24},{"id":"119b1ef3a45b55fde00887ae4db4b78a7fe7ec7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:37:25.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"b104843cce319d4003a191e22bd306e7db8c5df8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:39:27.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"36598132b28ccb2f58b464e99fcaa38396ad7518","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 15:43:47.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"a58629efdc95d0bf7fe740ba4ccf30c2a50f521c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 15:52:21.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09851030a691471586121990a47886e8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a64986320d2b442ab464a65c5894b426 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e66bab2651f4503a335de68cf26a0ff : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ee74be11c76d46b59c4920a686c92689 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/17a3736b8aa34cb597f3a8623d9a521e : SUCCESS in 8m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9b3154e833d3430b9aa726389d6f84f7 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":25},{"id":"6fcc75aad9f101a577d7d7a41236373c6a5508c9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 16:44:05.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"f56fa9388d140d877e26c789077cda9f155105fd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 16:56:51.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"854e726dcb040df0a22856942e0f55de7e898550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 17:58:39.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"7a2050b11dd591f77ce616d96ac18a4490bef9ff","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 17:58:42.000000000","message":"Patch Set 26:\n\n(4 comments)","accounts_in_message":[],"_revision_number":26},{"id":"c7f7a276646270eb1f9e5825c00f8b9eacebce6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 18:10:10.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb6392e5e2584f8c9ca84db8aeb97a84\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/891ea758f36449358d00bcc116f3477e : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e102cfe84453437484c0b86fb116084f : SUCCESS in 7m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5c52eee16ef84050a41e734d442be2f9 : SUCCESS in 9m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/216d5aaa3a4746b58c0c8c3e9d94880f : SUCCESS in 7m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a62777d37ead4cdfa4f0d85de1184976 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":26},{"id":"520558eade938d7be3460fefcde14deeadd87bd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 18:12:06.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"68f6d4d065dd706cfc6730da616e57ad5f386063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 18:13:09.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"1ffa97489594d8eaa3ca18b9d118deb06d832f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 18:35:29.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f222f034ef4185bea620cdef08ed61\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94a9fd51af5b440da11ed47d473f0b51 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d501ba679442495d8694f681476a3bf2 : SUCCESS in 5m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/07e6980f8000460e916ff092402bca5a : SUCCESS in 5m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/96119aacd1294b0d8a6d21a8123aa8e1 : SUCCESS in 8m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/57637b0c120947efba7b798737265519 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":28},{"id":"708b5835d0a067445dd04855a885ef7693c2b622","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 18:47:23.000000000","message":"Patch Set 28: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"8c25a35a0afe37ec96fe4501c4ccbae8f1f9c95c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:22:38.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"0d438694eccfb6f2e7ca1be7e9ea93e528ca46dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:23:21.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"82417780532f6b8d0d3b3e598f482bd4c27092dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:22:34.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8d0f29abae43f8b0544d0d8823f751\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/937f5e4a757b44ddb4d9657b86ef9b59 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2a44d90b7142e0801fe6be821c1273 : SUCCESS in 4m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/96f6b6b80d35410aa0032965fb2d4e44 : SUCCESS in 6m 24s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f5b289e401064cb1b8bd5a38dc06c229 : SUCCESS in 4m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b79af478263a40bd887691a4203e29c1 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":28},{"id":"3617d1842cdee144c7f07eca3d27dc58bb938416","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:22:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"6d317b4fccd9d5a1bf5005b576466aa13612ed5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:24:11.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ce0c83fe344ab683c99835617e9d48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d652aad1f614aea9e6b199836f67f17 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"234c0f72366585b36d607698875abd43c0f0d619","revisions":{"74db5821ad1a172dff3d42be8165d4bfad0d511f":{"kind":"REWORK","_number":1,"created":"2026-07-04 19:19:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/1"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/74db5821ad1a172dff3d42be8165d4bfad0d511f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/74db5821ad1a172dff3d42be8165d4bfad0d511f"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"90c33f44343ece953497d9a5c929105bb78999de":{"kind":"REWORK","_number":2,"created":"2026-07-04 19:31:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/2"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:31:51.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/90c33f44343ece953497d9a5c929105bb78999de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/90c33f44343ece953497d9a5c929105bb78999de"}]},"branch":"refs/heads/master"},"7603555b40f9c23870d75e304481d97526f38d8a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-04 19:32:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/3"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:32:19.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7603555b40f9c23870d75e304481d97526f38d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7603555b40f9c23870d75e304481d97526f38d8a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"56fa06957dbc9ccb5ad6b62390c4b45cc9441520":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-04 19:34:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/4"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:34:34.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56fa06957dbc9ccb5ad6b62390c4b45cc9441520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/56fa06957dbc9ccb5ad6b62390c4b45cc9441520"}]},"branch":"refs/heads/master","description":"Edit commit message"},"36edae76dd5f03af3f46b335a24dae74f9c2cee9":{"kind":"REWORK","_number":5,"created":"2026-07-05 17:15:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/5"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 17:15:58.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/36edae76dd5f03af3f46b335a24dae74f9c2cee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/36edae76dd5f03af3f46b335a24dae74f9c2cee9"}]},"branch":"refs/heads/master"},"b327af6f455c03e6ac60d1253c824baf1598ca2f":{"kind":"REWORK","_number":6,"created":"2026-07-05 17:22:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/6"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 17:22:52.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b327af6f455c03e6ac60d1253c824baf1598ca2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b327af6f455c03e6ac60d1253c824baf1598ca2f"}]},"branch":"refs/heads/master"},"4f0017d91afd5ea60fe10e37b129f36a15ab7258":{"kind":"REWORK","_number":7,"created":"2026-07-05 17:33:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/7"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 17:33:39.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f0017d91afd5ea60fe10e37b129f36a15ab7258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4f0017d91afd5ea60fe10e37b129f36a15ab7258"}]},"branch":"refs/heads/master"},"95f2e18e7f9cfcb53bd64c3812f5cb1c173b0991":{"kind":"REWORK","_number":8,"created":"2026-07-05 17:55:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/8"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 17:55:26.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95f2e18e7f9cfcb53bd64c3812f5cb1c173b0991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95f2e18e7f9cfcb53bd64c3812f5cb1c173b0991"}]},"branch":"refs/heads/master"},"0f29b04500508fc3417038ea70a75af4c16602d1":{"kind":"REWORK","_number":9,"created":"2026-07-05 18:43:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/9"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 18:43:43.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f29b04500508fc3417038ea70a75af4c16602d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f29b04500508fc3417038ea70a75af4c16602d1"}]},"branch":"refs/heads/master"},"3ef66e301156f5595cc096a61614746000feeba8":{"kind":"REWORK","_number":10,"created":"2026-07-05 18:51:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/10"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 18:51:45.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ef66e301156f5595cc096a61614746000feeba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ef66e301156f5595cc096a61614746000feeba8"}]},"branch":"refs/heads/master"},"162bd628ca7a9773fd526d5b6f6628f6ecee6d44":{"kind":"REWORK","_number":11,"created":"2026-07-05 18:53:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/11"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 18:53:57.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/162bd628ca7a9773fd526d5b6f6628f6ecee6d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/162bd628ca7a9773fd526d5b6f6628f6ecee6d44"}]},"branch":"refs/heads/master"},"7baa9332737ba93d569a7b16ea85c2548c72d151":{"kind":"REWORK","_number":12,"created":"2026-07-05 18:56:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/12"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 18:56:16.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7baa9332737ba93d569a7b16ea85c2548c72d151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7baa9332737ba93d569a7b16ea85c2548c72d151"}]},"branch":"refs/heads/master"},"baa1f50b6e0bb4315635268a884ec0113e1113f0":{"kind":"REWORK","_number":13,"created":"2026-07-05 19:03:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/13"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 19:03:58.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/baa1f50b6e0bb4315635268a884ec0113e1113f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/baa1f50b6e0bb4315635268a884ec0113e1113f0"}]},"branch":"refs/heads/master"},"460a8bdd5248b4277d8a30afc393c5178bb66fdd":{"kind":"REWORK","_number":14,"created":"2026-07-05 20:39:01.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/20/996020/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/14"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-05 20:36:24.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/460a8bdd5248b4277d8a30afc393c5178bb66fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/460a8bdd5248b4277d8a30afc393c5178bb66fdd"}]},"branch":"refs/heads/master"},"3fab6adc5d2d8995544f2b5e3db904a82a08009d":{"kind":"REWORK","_number":15,"created":"2026-07-05 20:46:45.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/20/996020/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/15"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-05 20:46:24.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fab6adc5d2d8995544f2b5e3db904a82a08009d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fab6adc5d2d8995544f2b5e3db904a82a08009d"}]},"branch":"refs/heads/master"},"8c305120d6458d26009fb2327e7eef8c53f038a1":{"kind":"REWORK","_number":16,"created":"2026-07-05 21:10:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/16"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 21:10:27.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8c305120d6458d26009fb2327e7eef8c53f038a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8c305120d6458d26009fb2327e7eef8c53f038a1"}]},"branch":"refs/heads/master"},"b703bc08c1cbef79a913097622d8e626da475e85":{"kind":"REWORK","_number":17,"created":"2026-07-05 21:18:15.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/20/996020/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/17"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-05 21:18:15.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b703bc08c1cbef79a913097622d8e626da475e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b703bc08c1cbef79a913097622d8e626da475e85"}]},"branch":"refs/heads/master"},"bcf14804543cb2f0bbb9e61cac3d3b4e9cd88edd":{"kind":"REWORK","_number":18,"created":"2026-07-05 21:48:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/18"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 21:48:41.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcf14804543cb2f0bbb9e61cac3d3b4e9cd88edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcf14804543cb2f0bbb9e61cac3d3b4e9cd88edd"}]},"branch":"refs/heads/master"},"6583a5c3f105a6965c19546437ce34c35402cb2d":{"kind":"REWORK","_number":19,"created":"2026-07-05 21:50:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/19"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-05 21:50:14.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6583a5c3f105a6965c19546437ce34c35402cb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6583a5c3f105a6965c19546437ce34c35402cb2d"}]},"branch":"refs/heads/master"},"b87317cca96ab78edc4271ce4d5794565b22f1ce":{"kind":"REWORK","_number":20,"created":"2026-07-06 11:43:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/20/996020/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/20"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 11:43:04.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b87317cca96ab78edc4271ce4d5794565b22f1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b87317cca96ab78edc4271ce4d5794565b22f1ce"}]},"branch":"refs/heads/master"},"b074d3221f04245e866332c4f886f76e53953abd":{"kind":"REWORK","_number":21,"created":"2026-07-06 14:53:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/21"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 14:53:55.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b074d3221f04245e866332c4f886f76e53953abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b074d3221f04245e866332c4f886f76e53953abd"}]},"branch":"refs/heads/master"},"54c01413cb63bb6f50c295705ab9f159a226b3c4":{"kind":"REWORK","_number":22,"created":"2026-07-06 14:55:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/22"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 14:55:04.000000000","tz":0},"subject":"WIP: Update Russian language","message":"WIP: Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54c01413cb63bb6f50c295705ab9f159a226b3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54c01413cb63bb6f50c295705ab9f159a226b3c4"}]},"branch":"refs/heads/master"},"3e468e0a3897f201e510716dce9a6d1c53d88a0e":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2026-07-06 14:56:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/23"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 14:56:56.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e468e0a3897f201e510716dce9a6d1c53d88a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e468e0a3897f201e510716dce9a6d1c53d88a0e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"167cf364c15f6e5807495c616920835a9922ca5f":{"kind":"REWORK","_number":24,"created":"2026-07-06 15:20:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/24"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 15:20:33.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/167cf364c15f6e5807495c616920835a9922ca5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/167cf364c15f6e5807495c616920835a9922ca5f"}]},"branch":"refs/heads/master"},"facd3716cc3280b2fe09b6a6f9a808d8a3f2cfe6":{"kind":"REWORK","_number":25,"created":"2026-07-06 15:37:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/25"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 15:37:25.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/facd3716cc3280b2fe09b6a6f9a808d8a3f2cfe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/facd3716cc3280b2fe09b6a6f9a808d8a3f2cfe6"}]},"branch":"refs/heads/master"},"12691a282ec2a04edf864a969d2ac3b53a938c29":{"kind":"REWORK","_number":26,"created":"2026-07-06 17:58:39.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/20/996020/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/26"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 17:57:43.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12691a282ec2a04edf864a969d2ac3b53a938c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12691a282ec2a04edf864a969d2ac3b53a938c29"}]},"branch":"refs/heads/master"},"73c3c3352bdf9fff10a748de7c7a204de12b005b":{"kind":"REWORK","_number":27,"created":"2026-07-06 18:12:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/27"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 18:12:05.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73c3c3352bdf9fff10a748de7c7a204de12b005b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73c3c3352bdf9fff10a748de7c7a204de12b005b"}]},"branch":"refs/heads/master"},"234c0f72366585b36d607698875abd43c0f0d619":{"kind":"REWORK","_number":28,"created":"2026-07-06 18:13:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/20/996020/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/996020/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/996020/28"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:19:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-06 18:13:08.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrases and fix mistakes\n\nChange-Id: I873ccd29c1c833eff3514531cf0908f7e51c2226\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/234c0f72366585b36d607698875abd43c0f0d619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/234c0f72366585b36d607698875abd43c0f0d619"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
