)]}'
{"id":"openstack%2Fskyline-console~996024","triplet_id":"openstack%2Fskyline-console~stable%2F2026.1~If034fa5afcfd2016057275f1b6d67a4e4b68824b","project":"openstack/skyline-console","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:19:33.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-07 09:30:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If034fa5afcfd2016057275f1b6d67a4e4b68824b","subject":"fix: Show red overflow state in quota chart for instance/volume creation","status":"MERGED","created":"2026-07-04 19:48:49.000000000","updated":"2026-07-07 09:31:01.000000000","submitted":"2026-07-07 09:30:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996024","cherry_pick_of_change":989929,"cherry_pick_of_patch_set":1,"meta_rev_id":"445d6676abd6d264a1988258a69ba016ed3693ac","_number":996024,"virtual_id_number":996024,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 09:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-07 03:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-07 03:19:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-04 19:48:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"CC"},{"updated":"2026-07-04 19:49:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-07-04 21:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 07:05:37.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"beb8dd57b9919b606334afb864f1a1df01108305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:48:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"895d7469b68289a97cef30a2040b0fb5cf39fe30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:49:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5466d60293c1e2e997a3cfa239d37b962ba2b52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 21:08:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b33d437f565f4cdebc2840bb9ca6516e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b68ce7c35c3d453c824c94027454db9a : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558509f340f240daa468253d2ee95e12 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ae47605d44624c85b610c6d373ccce41 : SUCCESS in 11m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/496ca45d7fb44e0b9163dbbcf793a248 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/695f9c79061a42d291290986dbf47637 : SUCCESS in 4m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/98644c58194541709c3fa918757f346a : SUCCESS in 53m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f06e526091f2485a87aace8f29063946 : SUCCESS in 29m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6e2e7f260c784e16969d09916b5c6de1 : FAILURE in 1h 14m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/862c63159d0740ba8b6f5e9a8f5feeb8 : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6457c95242964c268e3879586e8745607cbd4a75","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 21:29:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"daa6ea1448654a91aac85b5156e0e55938de4bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 22:33:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b066d0b15084ddf8dac809f45bd4788\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad974d58421a4d0ca6b1114004245ca7 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcf2f7b480649cc94e8c73fd9179057 : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/43f85a84c4b84d4aa59661cdd49c18d9 : SUCCESS in 5m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6d3945b0d08248329a02b7feacca5cbe : SUCCESS in 4m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ab9fdf4afadb479f9f8f0ba94f58ab40 : SUCCESS in 3m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9e951a8c27614d47a7ccc01528510645 : SUCCESS in 43m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c29143a6f1c444a3a1c789e878515f6f : SUCCESS in 59m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8459525c5d5d42dab41134d0517177ce : SUCCESS in 54m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f0314dbce42f45bba9549504a562b207 : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":2},{"id":"edc96e832ee4e603bd5af306e1799549a9c5a71d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:19:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d44aef7685aa316774aa5d2298d717da1a9c5a43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:20:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d608334b7871c42a32ebaaadbfc948404fd857d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:25:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c00c5102b61440f596c547f685ccf4c8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f40278c31ca4bc6ba2af24f41e9ccd2 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862cf11d476840dfaa105b65dfd85662 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/db5dc3268b51437795275570fa41e9ff : SUCCESS in 9m 53s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4cafa370060c494085cd2b03aa16af69 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7809f57851b94bcea61bf091b765bda5 : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f40237f97fc245f6bf90ccf85c670853 : SUCCESS in 40m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/99d41455052545c49ab985d7d1243eb0 : SUCCESS in 1h 00m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7fa432f7d96f44ffae4ddf95055d3311 : FAILURE in 42m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/68c706702e6a4693a6fec7f8bf3622e8 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8f9288a2454c1ce811564c607509f44a9d06264a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 07:05:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff0b257c064d70dbe3254499656ef914ee6a90c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 08:12:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3d24648ba94c77a229fba4f6bd75d3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a9e365d346a49e592e2cd4c1ab76426 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ecf84f778d4e3dac2ae811a3177177 : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/06a33c06b6764ae6846bb251884abc7b : SUCCESS in 10m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e1e3df0bb1354eff85b2ce30f6a7365d : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f24399fb2330434087dbfe0db04f849f : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/17ccea3f133c40ddb816bd7074deb847 : SUCCESS in 40m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fd86b5938d704d64a15d95db2d2b0c86 : SUCCESS in 58m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5a1ee310257a4608be00d5bdc44dd16b : SUCCESS in 43m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c573635edd594fdfafc7fead79f1cff0 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"52c3ce97c4a82ab649e32ff6813bafaa364d67af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 08:12:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e463a9d62142853c43f00d821fa661881c740a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 09:30:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b08a594bbf9c439da81a41a4f6b9d4b6\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4097a49eb4f3462eaea003e378ac4c81 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49972afd698480f8d2b4e6bba3ee850 : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fe444a2ec9c143c693ffc5fde96a17bf : SUCCESS in 12m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e15bc19ab2494116b9a9625f37c17bb4 : SUCCESS in 7m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/817fe013deff40c8a1b26004d1268321 : SUCCESS in 7m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f77f75f8cd514e55bfbf85bd48444d03 : SUCCESS in 1h 08m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1fea7d3f6a5d43fd82fcc50e29f9ad72 : SUCCESS in 1h 02m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/30f5c7026eab485c8529f005e39af7c4 : SUCCESS in 56m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2add56535f744441abb73db346351242 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4eac72eb3c1663ed7fd649bde2f78db23afcc79a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 09:30:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"445d6676abd6d264a1988258a69ba016ed3693ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 09:31:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2522105d4a654f38860ef2f16b8f550a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22efadb35b146ae80c11068fd1ec397 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7a9144b4ca4a32d09bbae54266565c86539b0ab","revisions":{"ea4817f0c58f6963776cd974de1f79c8488356a9":{"kind":"REWORK","_number":1,"created":"2026-07-04 19:48:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/996024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/24/996024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/1"}}},"commit":{"parents":[{"commit":"2403009dbf875fd6fbe1ed942c92f8d6abbeedff","subject":"Fix hypervisor display for dedicated CPUs (pCPU)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2403009dbf875fd6fbe1ed942c92f8d6abbeedff"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 06:43:37.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:48:49.000000000","tz":0},"subject":"fix: Show red overflow state in quota chart for instance/volume creation","message":"fix: Show red overflow state in quota chart for instance/volume creation\n\nFixed the quota visualization issue in instance and volume creation forms. When the requested resource exceeds the available quota, the quota chart will now display a red overflow indicator to clearly show the over-limit status, instead of showing incorrect or unchanged visuals.\n\nCloses-Bug: #2146904\nChange-Id: If034fa5afcfd2016057275f1b6d67a4e4b68824b\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n(cherry picked from commit 3f7723653e8af838b2a2cb258061fac9de595a56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea4817f0c58f6963776cd974de1f79c8488356a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea4817f0c58f6963776cd974de1f79c8488356a9"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"ad7fd9f2514be41fb43ab2d85f33f234ce22dc9b","ours":"2403009dbf875fd6fbe1ed942c92f8d6abbeedff","theirs":"3f7723653e8af838b2a2cb258061fac9de595a56","merge_strategy":"recursive","contains_conflicts":false}},"c7a9144b4ca4a32d09bbae54266565c86539b0ab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-04 19:49:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/996024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/24/996024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/24/996024/2"}}},"commit":{"parents":[{"commit":"f8860b9a931372a1f3666144b905bb33861ab33f","subject":"fix: Unify attach volume operations for instance list, detail page and architecture view","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f8860b9a931372a1f3666144b905bb33861ab33f"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-25 06:43:37.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:49:19.000000000","tz":0},"subject":"fix: Show red overflow state in quota chart for instance/volume creation","message":"fix: Show red overflow state in quota chart for instance/volume creation\n\nFixed the quota visualization issue in instance and volume creation forms. When the requested resource exceeds the available quota, the quota chart will now display a red overflow indicator to clearly show the over-limit status, instead of showing incorrect or unchanged visuals.\n\nCloses-Bug: #2146904\nChange-Id: If034fa5afcfd2016057275f1b6d67a4e4b68824b\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n(cherry picked from commit 3f7723653e8af838b2a2cb258061fac9de595a56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c7a9144b4ca4a32d09bbae54266565c86539b0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c7a9144b4ca4a32d09bbae54266565c86539b0ab"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"2403009dbf875fd6fbe1ed942c92f8d6abbeedff","ours":"ea4817f0c58f6963776cd974de1f79c8488356a9","theirs":"f8860b9a931372a1f3666144b905bb33861ab33f","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
