)]}'
{"id":"openstack%2Fskyline-console~996026","triplet_id":"openstack%2Fskyline-console~stable%2F2026.1~If053a63872123308bc1e054a4144cf597ca3a8ea","project":"openstack/skyline-console","branch":"stable/2026.1","topic":"bug-2113880-stable/2026.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:18:26.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-07 04:17:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If053a63872123308bc1e054a4144cf597ca3a8ea","subject":"fix: Empty Remote IP Prefix for remote group rules","status":"MERGED","created":"2026-07-04 19:52:16.000000000","updated":"2026-07-07 04:17:59.000000000","submitted":"2026-07-07 04:17:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996026-bug-2113880-stable/2026.1","cherry_pick_of_change":990049,"cherry_pick_of_patch_set":1,"meta_rev_id":"2e438db5b7c5395edfb1c97321b5c31653091ab7","_number":996026,"virtual_id_number":996026,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 04:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-07 03:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-07 03:18:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-04 19:52:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"CC"},{"updated":"2026-07-04 19:52:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-07-04 20:57:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a29c89ee4650b7599a96a8a74173d563f2461da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-04 19:52:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f1646318091817b4330df97277a16cca87990036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 20:57:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abedf50e51964a4fbec96251ebfc0ded\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0d5a7272f244b3187c07e92a3b641ef : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770dd221186b4e52befeed5cd5b5bfea : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0ac9561855bc40e1a90c3ebd1092d30d : SUCCESS in 11m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/07aad1f32ddf4b948a16c813a3d5933b : SUCCESS in 7m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9cc364af26a2439685fdb29f3d9f5369 : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4be50060cca64077a086c377149b37c0 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"41163cac5d59ff6dedde461d269eee3777e1c62d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:18:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c9340eaf6f4be5b2d4e5c48398adf0bd718bc1c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:18:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28a26f87d388dcbe4b0894116997ba551fcfb5f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:19:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b972d34c0ec133dea609bbf9f8b0117b36a098b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:17:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dcbe375e12b4775af137b9e4a1275ad\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5698524c79ed4d65a34c848f35402960 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5a1019d795406ba69b7cad129ac472 : SUCCESS in 6m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/775ecfd911af48a3b03ef7418d99a49f : SUCCESS in 5m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/de1abfa3a4c04147aaed2a0112895424 : SUCCESS in 5m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7ea641129a7a45d2a71e1640a20e1c04 : SUCCESS in 3m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/828738854ec74c3c910f1b4964a92b98 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2e438db5b7c5395edfb1c97321b5c31653091ab7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 04:17:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19ecb458c51cef75b08c10622b874ff448e0fb55","revisions":{"19ecb458c51cef75b08c10622b874ff448e0fb55":{"kind":"REWORK","_number":1,"created":"2026-07-04 19:52:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/26/996026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/996026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/996026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/996026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/996026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/996026/1"}}},"commit":{"parents":[{"commit":"2403009dbf875fd6fbe1ed942c92f8d6abbeedff","subject":"Fix hypervisor display for dedicated CPUs (pCPU)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2403009dbf875fd6fbe1ed942c92f8d6abbeedff"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-05-26 06:50:54.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-04 19:52:16.000000000","tz":0},"subject":"fix: Empty Remote IP Prefix for remote group rules","message":"fix: Empty Remote IP Prefix for remote group rules\n\nFix misleading display where full IP range was shown for rules bound to remote security group.\n\nCloses-Bug: #2113880\nChange-Id: If053a63872123308bc1e054a4144cf597ca3a8ea\nSigned-off-by: zhangjingwei \u003czhang.jingwei@algoblu.com\u003e\n(cherry picked from commit 5b71957314e205b39e6bf64597e1d6f2dc6fb697)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19ecb458c51cef75b08c10622b874ff448e0fb55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19ecb458c51cef75b08c10622b874ff448e0fb55"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"3f7723653e8af838b2a2cb258061fac9de595a56","ours":"2403009dbf875fd6fbe1ed942c92f8d6abbeedff","theirs":"5b71957314e205b39e6bf64597e1d6f2dc6fb697","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
