)]}'
{"id":"openstack%2Fskyline-console~996060","triplet_id":"openstack%2Fskyline-console~master~I637016177e65159574f19a0e0622ca935d7b9c74","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-07 05:19:29.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-07 03:19:56.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I637016177e65159574f19a0e0622ca935d7b9c74","subject":"Tighten instance card layout and set overflow-wrap: normal to stop label wrapping","status":"MERGED","created":"2026-07-06 07:35:19.000000000","updated":"2026-07-07 05:20:24.000000000","submitted":"2026-07-07 05:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996060","meta_rev_id":"350f97edb93c77c805a6f466ef5678c70ff08d03","_number":996060,"virtual_id_number":996060,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 05:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-07 03:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-06 14:44:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-07 03:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 09:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-06 14:44:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-06 14:44:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"72787afb16884327ca597ef53c38c2d5aa6724de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 07:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc41c1ee64f0a6c42192314e841e5b285a87270d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 08:00:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9d4ba496c7c6df79d33eccad279e820578548ed5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-06 08:52:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b4f8a49fc7686c854c89866254da42c518d6b9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 09:20:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abefcd2c0c34c189c0353a31afd1f33\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9b63eb63a94a5b833950c6d27b9bdf : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430293ccd8cb4b3db1f9b369d05b95c3 : SUCCESS in 5m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/80beac8bd97d4b2e96fdd99ac8fe0628 : SUCCESS in 10m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/39e3f9cac9e94494a87705a6bac92e46 : SUCCESS in 7m 46s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5119a05fb6894b1288c42d2fffb3cfbd : SUCCESS in 7m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9df4407621e54ffcaa1dabf050d0344a : SUCCESS in 1h 10m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/38f85ea646694d60b3e19e56b8b13ae1 : SUCCESS in 45m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5f5a544cb49244e4a57af349e28ecf39 : SUCCESS in 48m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8463ef7b8e1b4001ae2013dc7a2b0b0d : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b9f459af2981ecfd0f261b5492262bbc4796e0b5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-06 14:44:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5980dcaa07196e8cb52ca7091c4799c8358106df","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-07 03:19:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d43dbd3734a7149a137ddc7db22a1ce496f9eb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 03:20:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"826ad225382e99889a9b9ac756f9c37ba1fb3e79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:19:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6385cc9d80d24402bf37f1a79fe8f586\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36cb50909ee94a7fadc9979a4b17fcbb : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd30da7307845e781fc5d3e8bae483b : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c4bd9b7febed46938d834ce6be96da5e : SUCCESS in 5m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/99faf8bc6d164e82949470670efb093c : SUCCESS in 8m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d42ddfd6e8bf423086240edbc5d870b7 : SUCCESS in 5m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e163820561004efb8402069533a88d4d : SUCCESS in 1h 11m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cb096fd8a0954fbba45f84ae37fef8ea : SUCCESS in 34m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/73d98482eeaa4d42ad01fc77671026dd : SUCCESS in 28m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4ab565ed4fa94e96a55189d647e8e2bb : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6f5971a4847e319564319efb61e23bda56d1c3d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:19:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"350f97edb93c77c805a6f466ef5678c70ff08d03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 05:20:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703e8bc8b0c547beadabfd1935354884\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186185686df04221b54912303d002c0e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76ce3833b53c1f30581878083086eb0ef470c45e","revisions":{"fc8735b6a0b64b5456c3b8c8f4f5ce5f65298210":{"kind":"REWORK","_number":1,"created":"2026-07-06 07:35:19.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/60/996060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/60/996060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/1"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 07:35:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 07:35:19.000000000","tz":0},"subject":"wip:","message":"wip:\n\nChange-Id: I637016177e65159574f19a0e0622ca935d7b9c74\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fc8735b6a0b64b5456c3b8c8f4f5ce5f65298210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fc8735b6a0b64b5456c3b8c8f4f5ce5f65298210"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"76ce3833b53c1f30581878083086eb0ef470c45e":{"kind":"REWORK","_number":2,"created":"2026-07-06 08:00:10.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/60/996060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/60/996060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/60/996060/2"}}},"commit":{"parents":[{"commit":"5f92495047f253cf326fa4fad77a7324bbfdeaef","subject":"Add support for TOTP two-factor authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f92495047f253cf326fa4fad77a7324bbfdeaef"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 07:35:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-06 08:00:00.000000000","tz":0},"subject":"Tighten instance card layout and set overflow-wrap: normal to stop label wrapping","message":"Tighten instance card layout and set overflow-wrap: normal\nto stop label wrapping\n\nSet overflow-wrap: normal on labels (e.g Root disk)\nto prevent mid-word/phrase wrapping in another locales.\n\nChange-Id: I637016177e65159574f19a0e0622ca935d7b9c74\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/76ce3833b53c1f30581878083086eb0ef470c45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/76ce3833b53c1f30581878083086eb0ef470c45e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
