)]}'
{"id":"openstack%2Fskyline-console~996235","triplet_id":"openstack%2Fskyline-console~master~I516e7faac86b06a72e9bd53022594db465f9eef0","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-10 09:44:15.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-10 09:22:20.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-10 09:44:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I516e7faac86b06a72e9bd53022594db465f9eef0","subject":"Update Russian language","status":"MERGED","created":"2026-07-07 09:52:28.000000000","updated":"2026-07-10 09:45:17.000000000","submitted":"2026-07-10 09:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996235","meta_rev_id":"45e24f489dc5ecc2693e7e301841b861c1c579a4","_number":996235,"virtual_id_number":996235,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 09:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-10 09:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-07 10:24:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-10 09:22:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 10:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 10:24:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-07 10:24:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"c5e405eb23800b5aeb9df3f96be226fe04a2cf01","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 09:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"def462f64b160651bcdd6e56baaa9b96dd24a8a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 09:57:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"84cc1fa88040cc37f6f4975d65bf490ca8aca522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 10:18:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b881df40bdc421bbb0bf24e66b43044\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f8f56bf3185467c8e223771d02e5e25 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47dec28357b74664bbacc531f3de59ce : SUCCESS in 7m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1643f6b7745a4f8e84497c58412a3b7a : SUCCESS in 11m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9f23af47cd9041b0bf9751018ef64ef0 : SUCCESS in 8m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e6be587c314549569ec3799f564702ca : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7acc8196d0b511ca99d620c0129885b8b7720ac9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 10:24:08.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0e26ab89b2a146977edd1f7cb2313b751b27fbb5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 10:24:21.000000000","message":"Patch Set 3: Code-Review+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b594289f000fbb3d33a25f9b3c5140240efc363b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 10:43:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24f5180eda7242a6a6c6b3ea0e1d617f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e38ffedcb6a471f834091bec6b35b9e : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565358350fa2481cab0b6437d820c5a0 : SUCCESS in 6m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ca6e0b92e570465c8b9b545fb9abb506 : SUCCESS in 10m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1f488377a08e4191935be259c5c8deeb : SUCCESS in 9m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bbaade5ba0b647209c2ed01b67bcf4cf : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"63e8415dae3a1de8878a2ddeaa26eb049b7a306f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-10 09:22:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"906fa10d91eb0b895a11364ea2869a187ef73220","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-10 09:22:26.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d82f86b4bc43d8f1e68c45c797a4face75e33d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:22:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c2249d3d9e8885eaaf5dc4571e9c608d416acddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:44:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d73a318428457dae4d1e517f73af6f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93ae3870a09940b19e18bf2e54559969 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b17bacb7724e569c65bbdcee7d5f02 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a4939db7da2942fdbe6dbb571d866025 : SUCCESS in 11m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/00516bf871d847c49eeaf5f3980190be : SUCCESS in 8m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a86b8dd184e74b0283721f43a2e836ea : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":3},{"id":"312fa163eaf9babaad405258d4a7ba6a65f20493","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"45e24f489dc5ecc2693e7e301841b861c1c579a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:45:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b08d20692d23404aa1727b578639cbd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8349843b7b3456f9fd56a9cfff4e7c8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8928afa6a288da5240263f893c38f9d0d3ae4e73","revisions":{"76e12333ba7ef7d67ad17462aedeb2a476ec9720":{"kind":"REWORK","_number":1,"created":"2026-07-07 09:52:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/35/996235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/996235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/1"}}},"commit":{"parents":[{"commit":"1c8144122978a6bcbe49c88d71a086138b2c2f69","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c8144122978a6bcbe49c88d71a086138b2c2f69"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:52:28.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:52:28.000000000","tz":0},"subject":"Update Russian locales","message":"Update Russian locales\n\nChange-Id: I516e7faac86b06a72e9bd53022594db465f9eef0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/76e12333ba7ef7d67ad17462aedeb2a476ec9720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/76e12333ba7ef7d67ad17462aedeb2a476ec9720"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"8aede323cbd9adb5077d1abfe77f56c4ea21c3df":{"kind":"REWORK","_number":2,"created":"2026-07-07 09:57:41.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/35/996235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/996235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/2"}}},"commit":{"parents":[{"commit":"1c8144122978a6bcbe49c88d71a086138b2c2f69","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c8144122978a6bcbe49c88d71a086138b2c2f69"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:52:28.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:57:38.000000000","tz":0},"subject":"Update Russian locales","message":"Update Russian locales\n\nChange-Id: I516e7faac86b06a72e9bd53022594db465f9eef0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8aede323cbd9adb5077d1abfe77f56c4ea21c3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8aede323cbd9adb5077d1abfe77f56c4ea21c3df"}]},"branch":"refs/heads/master"},"8928afa6a288da5240263f893c38f9d0d3ae4e73":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-07 10:24:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/35/996235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/996235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/996235/3"}}},"commit":{"parents":[{"commit":"1c8144122978a6bcbe49c88d71a086138b2c2f69","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c8144122978a6bcbe49c88d71a086138b2c2f69"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:52:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-07 10:24:08.000000000","tz":0},"subject":"Update Russian language","message":"Update Russian language\n\nAdd new phrase\n\nChange-Id: I516e7faac86b06a72e9bd53022594db465f9eef0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8928afa6a288da5240263f893c38f9d0d3ae4e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8928afa6a288da5240263f893c38f9d0d3ae4e73"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
