)]}'
{"id":"openstack%2Fskyline-console~996237","triplet_id":"openstack%2Fskyline-console~master~I37412e8f0462cba4bbaf62583e8f7a4439ebd591","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-10 10:40:44.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-07-10 09:22:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I37412e8f0462cba4bbaf62583e8f7a4439ebd591","subject":"Explicitly define overflow for ant labels","status":"MERGED","created":"2026-07-07 09:59:55.000000000","updated":"2026-07-10 10:41:56.000000000","submitted":"2026-07-10 10:40:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996237","meta_rev_id":"84b519b1e3712a50bc1b76bf338217bd0de3a395","_number":996237,"virtual_id_number":996237,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 10:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-07-10 09:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-07 18:11:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-07-10 09:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 11:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 18:11:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-07 18:11:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"18d70299327b6c32de90264cf9dc4a94e2b1234f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 09:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59dc9f39dfb610b400668993a38ed38b53c135b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 10:45:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"17d162e4715ec02e35a08edca813ffc1b0c4634f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-07-07 10:45:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"947b68ad813f55bb5621d39ad21a23794d8720bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 11:54:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25ee5d44d5a4c0b87423f7c52814866\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74d32fb4a404850b9edc2934f36f5e0 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94de786d27b49d493139b7512065c7c : SUCCESS in 7m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bfb647703c6f459fa17b262b0e897e32 : SUCCESS in 10m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2225f4ec3e1646248726091b86897120 : SUCCESS in 7m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7c0aa143c3b8410ba168ad2223841067 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/420480faec584762ab9e99ea3ea660e6 : SUCCESS in 1h 08m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a8f3c29369f74267a90ed6db8e9d56be : SUCCESS in 1h 04m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/148d8a8f81e049e6b0596dc49a2cc759 : SUCCESS in 1h 04m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ab902fddea184b6f86c595cfa37aa05c : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"07e3da972ceb6a22b7dc830c10b0202a6d675b64","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-07 18:11:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e6683e62446639c12b8cc3c246e956f754bcd3fd","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-10 09:22:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3f4625ce78af7590e609586082f13de74ce652f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-07-10 09:22:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b68a3967a7c5833e76e6c02f4d174e1879aa695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:23:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b802b8a5e1dc6c8c882e53a7f1ea5403e2fc441f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 10:40:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c80eb10156444e9f4a33f7c1ea8807\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ff1acf38614bbb8fdbab7ebc32c668 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40d73839ed7408984129ac03bea1f40 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/62d01f8f553c45d7868cfea1ae3ecf50 : SUCCESS in 11m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9e181bc55aa24ca2806b960fc2082b32 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5e2bdc0c809043e4aefc77f690074dab : SUCCESS in 9m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cccac2189cbb41a3b9450fec4127e992 : SUCCESS in 1h 11m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/61a9b2fb24e441b18b93e05eaaf17d5f : SUCCESS in 1h 05m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d4c3f56bc64a404197c4e3f619a56e42 : SUCCESS in 55m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a1c7015eb2fb4e75921cd1ba6af71dad : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5cba3084791579b38f02fe8f958e4e46e784b595","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 10:40:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"84b519b1e3712a50bc1b76bf338217bd0de3a395","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 10:41:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed620eb346644c6989dca65f498a9a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b01e93b4a284e63a462d6b66aa8cadc : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e75581ca7747f6002f710328e142b65aa7afcac3","revisions":{"97aeda5e8b6e645489f3855d07e396aa72277d25":{"kind":"REWORK","_number":1,"created":"2026-07-07 09:59:55.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/37/996237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/996237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/1"}}},"commit":{"parents":[{"commit":"1c8144122978a6bcbe49c88d71a086138b2c2f69","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c8144122978a6bcbe49c88d71a086138b2c2f69"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:59:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:59:55.000000000","tz":0},"subject":"overflow","message":"overflow\n\nChange-Id: I37412e8f0462cba4bbaf62583e8f7a4439ebd591\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/97aeda5e8b6e645489f3855d07e396aa72277d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/97aeda5e8b6e645489f3855d07e396aa72277d25"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"e75581ca7747f6002f710328e142b65aa7afcac3":{"kind":"REWORK","_number":2,"created":"2026-07-07 10:45:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/37/996237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/37/996237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/37/996237/2"}}},"commit":{"parents":[{"commit":"1c8144122978a6bcbe49c88d71a086138b2c2f69","subject":"Merge \"Update Russian language\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c8144122978a6bcbe49c88d71a086138b2c2f69"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 09:59:55.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-07-07 10:45:38.000000000","tz":0},"subject":"Explicitly define overflow for ant labels","message":"Explicitly define overflow for ant labels\n\nExplicitly set overflow: visible for .ant-form-item-label to\noverride antd\u0027s default overflow\n\nChange-Id: I37412e8f0462cba4bbaf62583e8f7a4439ebd591\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e75581ca7747f6002f710328e142b65aa7afcac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e75581ca7747f6002f710328e142b65aa7afcac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
