)]}'
{"id":"openstack%2Fsolum~912675","triplet_id":"openstack%2Fsolum~stable%2Fxena~I2d5e177ee3a89e0cedd79e564505e6d41f77b8e4","project":"openstack/solum","branch":"stable/xena","topic":"drop-centos-7","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-03-13 15:58:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d5e177ee3a89e0cedd79e564505e6d41f77b8e4","subject":"Zuul: Remove references to centos-7 job nodes","status":"MERGED","created":"2024-03-12 22:53:03.000000000","updated":"2024-03-13 16:04:12.000000000","submitted":"2024-03-13 15:58:47.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912675-drop-centos-7","meta_rev_id":"2a808f4d1295f45116f3c96e01019840600d779b","_number":912675,"virtual_id_number":912675,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-03-13 15:58:47.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-12 23:07:35.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-03-13 15:58:47.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":1,"date":"2024-03-13 15:58:47.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 23:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 15:58:47.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"e6976d9e8865b82cf7ed4f5dd64b17f084ab8a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-03-12 22:53:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ccf525ee50b610434d6e54f5fe00b7b1db128e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 23:07:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c320e155a5db47ef9167b0f817348e6f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4aef98c393c4e2ca4b582b808cb63a4 : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/530adf5242544ba6928d50e22499ffb9 : FAILURE in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5cb5dbf194466a9720df0087e57165 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d255f57ffce4a56b8a4ce72b2dc101e : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bcd85e677c64c8c9be7944f4b6e3c4e : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08ac66e113a34040b0ed4b7afa1865fd : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cd2d8accff457b98a0a8eaa2056c62 : SUCCESS in 5m 50s\n- solum-devstack https://zuul.opendev.org/t/openstack/build/4934a120f3f74ea581724cd2a27c8734 : FAILURE in 10m 20s\n- solum-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95bf526d188b4e0bb7c67b1c09131ed4 : FAILURE in 9m 17s\n- solum-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba1776bcede24f1c8b908cd41158bfb9 : FAILURE in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd5075338c95a20737c014678ad4dd19894a83cc","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-03-13 15:58:47.000000000","message":"Patch Set 1: Code-Review+2 Verified+2 Workflow+1\n\nBypassing Zuul to merge this trivial cleanup after consulting project leadership.","accounts_in_message":[],"_revision_number":1},{"id":"36d0a8472e34871882332da18999d98021fad2e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-03-13 15:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a808f4d1295f45116f3c96e01019840600d779b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:04:12.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1310c970e17f40d980892d9c92095bd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633b900914044aa483e1436312617491 : FAILURE in 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"557d67c02f09384eb4c0d9d45bf52dcb08c3e2ba","revisions":{"557d67c02f09384eb4c0d9d45bf52dcb08c3e2ba":{"kind":"REWORK","_number":1,"created":"2024-03-12 22:53:03.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/75/912675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/solum","ref":"refs/changes/75/912675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/solum refs/changes/75/912675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/solum refs/changes/75/912675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/solum refs/changes/75/912675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/solum refs/changes/75/912675/1"}}},"commit":{"parents":[{"commit":"b5c93996fe756dfaa893800c741b4ca00bd40c5f","subject":"Fix deprecated zuul queue syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/solum/commit/b5c93996fe756dfaa893800c741b4ca00bd40c5f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-12 22:48:12.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-03-12 22:52:51.000000000","tz":0},"subject":"Zuul: Remove references to centos-7 job nodes","message":"Zuul: Remove references to centos-7 job nodes\n\nOpenDev is preparing to remove centos-7 nodes on March 15[*]. This\nchange drops Zuul configuration references to anything directly or\nindirectly relying on that label.\n\n[*] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/A2YIY5L7MVYSQMTVZU3L3OM7GLVVZPLK/\n\nChange-Id: I2d5e177ee3a89e0cedd79e564505e6d41f77b8e4\n(cherry picked from commit f6640157df6d384e8eab25291319ceb5df483546)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/solum/commit/557d67c02f09384eb4c0d9d45bf52dcb08c3e2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/solum/commit/557d67c02f09384eb4c0d9d45bf52dcb08c3e2ba"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
