)]}'
{"id":"openstack%2Fspecs-cookiecutter~732365","triplet_id":"openstack%2Fspecs-cookiecutter~master~I4e5abc8485103f5b830759056db24185670c5ec2","project":"openstack/specs-cookiecutter","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"I4e5abc8485103f5b830759056db24185670c5ec2","subject":"Update to openstackdocstheme","status":"MERGED","created":"2020-06-01 13:47:23.000000000","updated":"2020-06-09 18:41:08.000000000","submitted":"2020-06-09 18:41:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732365-1591728068957-1e1c92d6","meta_rev_id":"09ba89f99f7096ac22cf7bf1c7829e5e1d498a49","_number":732365,"virtual_id_number":732365,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 18:41:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-09 18:34:37.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-09 18:40:56.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-09 18:40:56.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 10:03:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-09 18:34:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-09 18:40:56.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-06-09 18:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ac4c651fd6505994277b1647d546eda78f076c4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-01 13:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1030c631965b9bf47548826c05ffbf52b9f8f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/f81d618a29b043899fea3b66d232e211 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"5ed4abfab21da1f5d67e5d2ce1ef95637edbe1f2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 10:03:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"972add53caaeff436d6d20c3a4740aed65d21fb9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 07:04:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8c4ade4e977e36e032145c5a187c95fdcc2d735c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 07:05:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/023a6e2fec984f03af70471a4d7c2620 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4553abbcd568edd74aa516d54bd4b6cababaecfc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 18:34:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"60f533de8c4b8c45e798e0a65a9bbf97d7d9f173","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-06-09 18:40:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5dc1cea12232a4c2087735ee2719d42efe88d410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:41:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"946cdfdee2475493d97538f37947f0096f5ea1e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f62b3c6968f9f214c4bb364d117f2c6bba99d04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:41:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/d0f066ba362a443e8d01c3e42a911c81 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9238dbe1c09bcfeb14c0577ba13d3e64b6eba82d","revisions":{"32013a2bcb900926a0e348489a2cc3b008e2d36a":{"kind":"REWORK","_number":1,"created":"2020-06-01 13:47:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/65/732365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/specs-cookiecutter","ref":"refs/changes/65/732365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/1"}}},"commit":{"parents":[{"commit":"ae6b370636bf44407b60cb137c702ab2561dfac5","subject":"Fix gitreview host name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/ae6b370636bf44407b60cb137c702ab2561dfac5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-01 13:45:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-01 13:46:20.000000000","tz":120},"subject":"Update to openstackdocstheme","message":"Update to openstackdocstheme\n\noslosphinx is dead, convert to openstackdocstheme and update for\npython3.\n\nChange-Id: I4e5abc8485103f5b830759056db24185670c5ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/32013a2bcb900926a0e348489a2cc3b008e2d36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/32013a2bcb900926a0e348489a2cc3b008e2d36a"}]},"branch":"refs/heads/master"},"9238dbe1c09bcfeb14c0577ba13d3e64b6eba82d":{"kind":"REWORK","_number":2,"created":"2020-06-09 07:04:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/65/732365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/specs-cookiecutter","ref":"refs/changes/65/732365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/specs-cookiecutter refs/changes/65/732365/2"}}},"commit":{"parents":[{"commit":"ae6b370636bf44407b60cb137c702ab2561dfac5","subject":"Fix gitreview host name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/ae6b370636bf44407b60cb137c702ab2561dfac5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-01 13:45:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-09 07:04:43.000000000","tz":0},"subject":"Update to openstackdocstheme","message":"Update to openstackdocstheme\n\noslosphinx is dead, convert to openstackdocstheme and update for\npython3.\n\nChange-Id: I4e5abc8485103f5b830759056db24185670c5ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/9238dbe1c09bcfeb14c0577ba13d3e64b6eba82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/specs-cookiecutter/commit/9238dbe1c09bcfeb14c0577ba13d3e64b6eba82d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
