)]}'
{"id":"openstack%2Fstackviz~1007966","triplet_id":"openstack%2Fstackviz~master~Iee703bedf030a74723b0b6fb094fd054a7ede0b4","project":"openstack/stackviz","branch":"master","topic":"2027.1-testing","attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-29 22:47:27.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iee703bedf030a74723b0b6fb094fd054a7ede0b4","subject":"Update testing as per the latest testing runtime","status":"NEW","created":"2026-09-29 21:19:58.000000000","updated":"2026-09-30 00:15:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"80369bbc4e170697b9dc7901e1bfcd9dbafd1776","_number":1007966,"virtual_id_number":1007966,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-29 22:47:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-29 22:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cce73f72c82f0021d0858badf1a0d84742e0987b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-29 21:19:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70b13aa8ec0e4fb80442cdd5914f7ea17acafdd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:47:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/719f378bf94446bca9b597bca6c13e5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b94887bd2f432fb0f7573a63da5729 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4102711f24cf4895a379b422fa9b81e4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40fc7f3dbddf48afb4d6509fda6d7ac6 : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baf41a151325465c95c844360b1e2ad7 : FAILURE in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1594cfba08a454b8ee9abe5ce0bad59 : FAILURE in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36f595ace3a644b6980bb5c6a3881901 : FAILURE in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de4d9b09306f438cad22dc2a45605a0c : FAILURE in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ffeb9836b4045859d8905a3f03cf39c : FAILURE in 8m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7ce4d03f0bf4653afd0679216564d7c : FAILURE in 5m 36s\n- stackviz-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d58763b0f31941329d48ca69e2ce7c21 : SUCCESS in 6m 00s\n- stackviz-nodejs10-run-test-browser https://zuul.opendev.org/t/openstack/build/ff7e30b2b8f54898855d2a2be3ca49f7 : SUCCESS in 11m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d1a026f89fa42a2becee10e3e80b227 : SUCCESS in 1h 24m 15s\n- tempest-full-oldest-stable https://zuul.opendev.org/t/openstack/build/d3e8dffecc8846b682f23a6eb0972277 : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bc0f4992511c6cb4c2d6dcceea1f42789ee62aeb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-30 00:15:03.000000000","message":"Topic qa-2027-1 removed","accounts_in_message":[],"_revision_number":1},{"id":"80369bbc4e170697b9dc7901e1bfcd9dbafd1776","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-30 00:15:07.000000000","message":"Topic set to 2027.1-testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8dc19494e46a5d7236b891ad6ab2de190add73d","revisions":{"a8dc19494e46a5d7236b891ad6ab2de190add73d":{"kind":"REWORK","_number":1,"created":"2026-09-29 21:19:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/1007966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stackviz","ref":"refs/changes/66/1007966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stackviz refs/changes/66/1007966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stackviz refs/changes/66/1007966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stackviz refs/changes/66/1007966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stackviz refs/changes/66/1007966/1"}}},"commit":{"parents":[{"commit":"45de3fd70db2c05ffee48f515ee995481da2ce2e","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stackviz/commit/45de3fd70db2c05ffee48f515ee995481da2ce2e"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-29 21:14:54.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-29 21:19:57.000000000","tz":-420},"subject":"Update testing as per the latest testing runtime","message":"Update testing as per the latest testing runtime\n\nChange-Id: Iee703bedf030a74723b0b6fb094fd054a7ede0b4\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stackviz/commit/a8dc19494e46a5d7236b891ad6ab2de190add73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stackviz/commit/a8dc19494e46a5d7236b891ad6ab2de190add73d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
