)]}'
{"id":"openstack%2Fstevedore~652818","triplet_id":"openstack%2Fstevedore~master~Ib38da77120fd0392984388350bb785eae5893c06","project":"openstack/stevedore","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Ib38da77120fd0392984388350bb785eae5893c06","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-16 04:43:59.000000000","updated":"2019-04-24 14:27:33.000000000","submitted":"2019-04-24 14:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652818-1556116053421-a380670c","meta_rev_id":"f5288e1a40e851861e58e3bea91f94064bc9e260","_number":652818,"virtual_id_number":652818,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-04-24 14:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-04-24 12:46:40.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-04-16 13:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-04-24 12:46:40.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 13:29:53.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-04-24 12:46:40.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-04-24 14:27:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c55d8e416085dde923ddfda9b6faddfd6562431","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-16 04:43:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa9e60cde3c8bbc486cf13933241f6ce683ecc2d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-16 04:50:34.000000000","message":"Topic set to dropping-py35-testing","accounts_in_message":[],"_revision_number":1},{"id":"82ea03595e890ea7172eee6859fe3f674fd95a9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:36:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/652818/1/check/test-release-openstack/56bf2ee/ : SUCCESS in 3m 28s\n- tempest-full http://logs.openstack.org/18/652818/1/check/tempest-full/0f7548e/ : SUCCESS in 1h 46m 01s\n- tempest-full-py3 http://logs.openstack.org/18/652818/1/check/tempest-full-py3/7107e96/ : SUCCESS in 1h 33m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652818/1/check/openstack-tox-lower-constraints/4b1ca56/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/18/652818/1/check/openstack-tox-pep8/6b83ebd/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/18/652818/1/check/openstack-tox-py27/fe0b874/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/18/652818/1/check/openstack-tox-py36/dd9a224/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/18/652818/1/check/openstack-tox-py37/3a62961/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/18/652818/1/check/openstack-tox-docs/4c90dc9/html/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/18/652818/1/check/build-openstack-releasenotes/5159f8a/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5cb5b297482254f59bedaf4d04a9ab814fd9b533","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-04-16 13:29:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2b0a6246828220ddc418509614b333b0ab31162","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-04-24 12:46:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f13e40817bba9a06728308dd8cdb6be750cb7e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:46:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfc9b2c6a6e067637060a0022f178e104ada326c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:27:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f5288e1a40e851861e58e3bea91f94064bc9e260","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:27:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/652818/1/gate/test-release-openstack/023d2b8/ : SUCCESS in 3m 13s\n- tempest-full http://logs.openstack.org/18/652818/1/gate/tempest-full/f774ba7/ : SUCCESS in 1h 38m 21s\n- tempest-full-py3 http://logs.openstack.org/18/652818/1/gate/tempest-full-py3/30137a9/ : SUCCESS in 1h 25m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652818/1/gate/openstack-tox-lower-constraints/0badc48/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/18/652818/1/gate/openstack-tox-pep8/11ec480/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/18/652818/1/gate/openstack-tox-py27/07200f4/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/18/652818/1/gate/openstack-tox-py36/9d4aa47/ : SUCCESS in 7m 58s\n- openstack-tox-py37 http://logs.openstack.org/18/652818/1/gate/openstack-tox-py37/fd697c3/ : SUCCESS in 3m 50s\n- openstack-tox-docs http://logs.openstack.org/18/652818/1/gate/openstack-tox-docs/27786df/html/ : SUCCESS in 3m 27s\n- build-openstack-releasenotes http://logs.openstack.org/18/652818/1/gate/build-openstack-releasenotes/5608afa/html/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"936f18e05b6b66bf8e288b7f73f08d163b7e2eb5","revisions":{"936f18e05b6b66bf8e288b7f73f08d163b7e2eb5":{"kind":"REWORK","_number":1,"created":"2019-04-16 04:43:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/652818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/18/652818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/18/652818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/18/652818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/18/652818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/18/652818/1"}}},"commit":{"parents":[{"commit":"9de37d322732a0157720a471786f9581f1878c50","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9de37d322732a0157720a471786f9581f1878c50"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-16 04:43:40.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-16 04:43:57.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ib38da77120fd0392984388350bb785eae5893c06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/936f18e05b6b66bf8e288b7f73f08d163b7e2eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/936f18e05b6b66bf8e288b7f73f08d163b7e2eb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
