)]}'
{"id":"openstack%2Fstevedore~655304","triplet_id":"openstack%2Fstevedore~master~Ie3617d76b6131386c2fa241b41888436ed44909f","project":"openstack/stevedore","branch":"master","hashtags":[],"change_id":"Ie3617d76b6131386c2fa241b41888436ed44909f","subject":"update git.openstack.org to opendev","status":"MERGED","created":"2019-04-24 06:15:28.000000000","updated":"2019-05-03 18:57:17.000000000","submitted":"2019-05-03 18:57:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655304-1556909837797-172fa3e6","meta_rev_id":"e9d3d1299fb71c2fa0c14976d63de180dc213752","_number":655304,"virtual_id_number":655304,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-04-25 01:07:57.000000000","_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-03 18:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 16:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},{"value":1,"date":"2019-05-03 16:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 16:40:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 16:55:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-03 18:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc05d7fa281ad0f414240858cbec5cff6bb025f5","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 06:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e099d3ddc8d8f0ee6f7048a99127c695ec2e5cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/04/655304/1/check/test-release-openstack/2185850/ : SUCCESS in 3m 19s\n- tempest-full http://logs.openstack.org/04/655304/1/check/tempest-full/46b9ed2/ : SUCCESS in 1h 45m 08s\n- tempest-full-py3 http://logs.openstack.org/04/655304/1/check/tempest-full-py3/ceb3056/ : SUCCESS in 1h 38m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/655304/1/check/openstack-tox-lower-constraints/66e8b06/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/04/655304/1/check/openstack-tox-pep8/96bea0d/ : SUCCESS in 10m 14s\n- openstack-tox-py27 http://logs.openstack.org/04/655304/1/check/openstack-tox-py27/349c293/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/04/655304/1/check/openstack-tox-py35/5fd5468/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/04/655304/1/check/openstack-tox-py36/85f67e8/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/04/655304/1/check/openstack-tox-py37/0dca8c0/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/04/655304/1/check/openstack-tox-docs/027f6b8/html/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/04/655304/1/check/build-openstack-releasenotes/ef51861/html/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9c3bfec2acbcb886296f601f39229d65df0f14fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-24 10:01:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c2522bd829d2506059ccb9a29a98d9deae1e19dc","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-25 01:07:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8bc59daef070034593090902b7aadbfdeabc120d","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-25 01:07:57.000000000","message":"Patch Set 2:\n\n\u003e (3 comments)\n\nyes,you are right.thank you.","accounts_in_message":[],"_revision_number":2},{"id":"23b7b655d26d6aafb247b9382eaccb5044c1485e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 03:14:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/04/655304/2/check/test-release-openstack/66d8aa3/ : SUCCESS in 3m 14s\n- tempest-full http://logs.openstack.org/04/655304/2/check/tempest-full/1323eda/ : SUCCESS in 1h 35m 05s\n- tempest-full-py3 http://logs.openstack.org/04/655304/2/check/tempest-full-py3/5de300b/ : SUCCESS in 1h 36m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/655304/2/check/openstack-tox-lower-constraints/db24fe2/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/04/655304/2/check/openstack-tox-pep8/dfcc0eb/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/04/655304/2/check/openstack-tox-py27/a7d7d0a/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/04/655304/2/check/openstack-tox-py36/3b5e754/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/04/655304/2/check/openstack-tox-py37/b57a91e/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/04/655304/2/check/openstack-tox-docs/a0d2745/html/ : SUCCESS in 4m 12s\n- build-openstack-releasenotes http://logs.openstack.org/04/655304/2/check/build-openstack-releasenotes/ddbe3f7/html/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"59c8305a85526c71827967bef3cea5150ac98abc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:40:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c6d84f574596a7322c2436992df1ac85920cb29","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-03 16:55:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a8a90af65151db516f3af48b3da4fdca031ada1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:55:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc2098cfbf621728d350dbebda12b0027b999405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:57:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/04/655304/2/gate/test-release-openstack/b764e4a/ : SUCCESS in 3m 00s\n- tempest-full http://logs.openstack.org/04/655304/2/gate/tempest-full/5342d09/ : SUCCESS in 1h 49m 11s\n- tempest-full-py3 http://logs.openstack.org/04/655304/2/gate/tempest-full-py3/2a3883b/ : SUCCESS in 1h 48m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/655304/2/gate/openstack-tox-lower-constraints/e3764f0/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/04/655304/2/gate/openstack-tox-pep8/331f5ef/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/04/655304/2/gate/openstack-tox-py27/b6fc298/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/04/655304/2/gate/openstack-tox-py36/406c487/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/04/655304/2/gate/openstack-tox-py37/d7ad092/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/04/655304/2/gate/openstack-tox-docs/27dfe2b/html/ : SUCCESS in 5m 27s\n- build-openstack-releasenotes http://logs.openstack.org/04/655304/2/gate/build-openstack-releasenotes/81b6694/html/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e9d3d1299fb71c2fa0c14976d63de180dc213752","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:57:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b13d7fe5998792ce9abba8429a1161e7939e2f3c","revisions":{"adfc7fb52363c1f83b6dfb06a47b8d2d6cc0c946":{"kind":"REWORK","_number":1,"created":"2019-04-24 06:15:28.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/04/655304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/04/655304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/04/655304/1"}}},"commit":{"parents":[{"commit":"4a25aa10cf3810c0e03cce6240309445d0a96317","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4a25aa10cf3810c0e03cce6240309445d0a96317"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:14:57.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:15:24.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Ie3617d76b6131386c2fa241b41888436ed44909f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/adfc7fb52363c1f83b6dfb06a47b8d2d6cc0c946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/adfc7fb52363c1f83b6dfb06a47b8d2d6cc0c946"}]},"branch":"refs/heads/master"},"b13d7fe5998792ce9abba8429a1161e7939e2f3c":{"kind":"REWORK","_number":2,"created":"2019-04-25 01:07:20.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/04/655304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/04/655304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/04/655304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/04/655304/2"}}},"commit":{"parents":[{"commit":"4a25aa10cf3810c0e03cce6240309445d0a96317","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/4a25aa10cf3810c0e03cce6240309445d0a96317"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 06:14:57.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-25 01:07:11.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: Ie3617d76b6131386c2fa241b41888436ed44909f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/b13d7fe5998792ce9abba8429a1161e7939e2f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/b13d7fe5998792ce9abba8429a1161e7939e2f3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
