)]}'
{"id":"openstack%2Fstevedore~659551","triplet_id":"openstack%2Fstevedore~master~Ieaf9c77d62551352833a644ab22dff261c08bb78","project":"openstack/stevedore","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"Ieaf9c77d62551352833a644ab22dff261c08bb78","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-16 12:41:16.000000000","updated":"2019-05-17 11:28:52.000000000","submitted":"2019-05-17 11:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659551-1558092532042-61dee02a","meta_rev_id":"1f5b9f438e5e4e2d3cd3ab3633a6359711801198","_number":659551,"virtual_id_number":659551,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-05-17 07:33:40.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-17 11:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-16 15:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-16 15:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 15:30:54.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-17 11:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58a01d8cf30fb07de4d9baa46de245b3236e79fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-16 12:41:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"038829587f26fe904ebb2153baf7757e9df9b8e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-16 12:43:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c59dfeaed094cd43279a80e32fae1ad5bfb3e42e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-16 13:43:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWhoops, I missed that this was an Oslo lib too. Thanks for fixing it.","accounts_in_message":[],"_revision_number":2},{"id":"a7467670808ca09a38d101f39e83ae3d05e7b244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:20:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/51/659551/2/check/requirements-check/acfbd71/ : FAILURE in 4m 18s\n- tempest-full http://logs.openstack.org/51/659551/2/check/tempest-full/100d707/ : SUCCESS in 1h 36m 01s\n- tempest-full-py3 http://logs.openstack.org/51/659551/2/check/tempest-full-py3/e346131/ : SUCCESS in 1h 32m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/2/check/openstack-tox-lower-constraints/cc7a74f/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/2/check/openstack-tox-pep8/1651769/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/2/check/openstack-tox-py27/c3cb332/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/2/check/openstack-tox-py36/6b3941f/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/2/check/openstack-tox-py37/923f4bd/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/51/659551/2/check/openstack-tox-docs/d4fd353/html/ : SUCCESS in 5m 30s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/2/check/build-openstack-releasenotes/d63afcc/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e193448ce024769b508fb5914047d2e345ae2fe1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-16 14:30:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b18e7de61e6be2c434ed808617377a9637e8e969","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-16 15:30:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nOops, I missed the lower constraint change here too. :-)","accounts_in_message":[],"_revision_number":3},{"id":"e15f16296ef39eda63a8d2e1267d3a1e7adace06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:21:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/51/659551/3/check/requirements-check/93a1bce/ : SUCCESS in 3m 56s\n- tempest-full http://logs.openstack.org/51/659551/3/check/tempest-full/c0685eb/ : SUCCESS in 1h 49m 15s\n- tempest-full-py3 http://logs.openstack.org/51/659551/3/check/tempest-full-py3/9e9a87d/ : SUCCESS in 1h 44m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/3/check/openstack-tox-lower-constraints/a4ae773/ : SUCCESS in 10m 28s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/3/check/openstack-tox-pep8/dd46d97/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/3/check/openstack-tox-py27/52e4e97/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/3/check/openstack-tox-py36/de71a5b/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/3/check/openstack-tox-py37/68c05b6/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/51/659551/3/check/openstack-tox-docs/254cdc3/html/ : SUCCESS in 5m 10s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/3/check/build-openstack-releasenotes/4971f30/html/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"edc99c972dbb3d5bdb3034f4fef569c40f32f635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:21:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"936814cf3269f1e764e80c8a7d2612f272578245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:30:25.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"add8b5c4a63d2674f45ee3f2b768e5302ab38ac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 18:36:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/51/659551/3/gate/requirements-check/5803612/ : SUCCESS in 4m 43s\n- tempest-full http://logs.openstack.org/51/659551/3/gate/tempest-full/8fe1474/ : TIMED_OUT in 2h 05m 10s\n- tempest-full-py3 http://logs.openstack.org/51/659551/3/gate/tempest-full-py3/4c0c3c1/ : SUCCESS in 1h 41m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/3/gate/openstack-tox-lower-constraints/de6ec2a/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/3/gate/openstack-tox-pep8/7c8dfb5/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py27/2f9a846/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py36/3efd657/ : SUCCESS in 6m 36s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py37/32cd935/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/51/659551/3/gate/openstack-tox-docs/a70e3a1/html/ : SUCCESS in 5m 49s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/3/gate/build-openstack-releasenotes/ac25de8/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b4a56db18a704e781efaa930d00ca0a675921381","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-16 18:40:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a12ab361bfbdf39964d7b575fbd41f942b430c56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 20:51:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/51/659551/3/check/requirements-check/7e7a517/ : SUCCESS in 4m 23s\n- tempest-full http://logs.openstack.org/51/659551/3/check/tempest-full/6d7262d/ : TIMED_OUT in 2h 07m 37s\n- tempest-full-py3 http://logs.openstack.org/51/659551/3/check/tempest-full-py3/70925b4/ : SUCCESS in 1h 34m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/3/check/openstack-tox-lower-constraints/9f4d2b5/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/3/check/openstack-tox-pep8/ba7f4f3/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/3/check/openstack-tox-py27/8d49a34/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/3/check/openstack-tox-py36/1b824b0/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/3/check/openstack-tox-py37/24bfbb7/ : SUCCESS in 6m 28s\n- openstack-tox-docs http://logs.openstack.org/51/659551/3/check/openstack-tox-docs/5ff439a/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/3/check/build-openstack-releasenotes/89a65f6/html/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"066c900439d0c62ce73922d3ff64d2335f7b82ba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-17 07:33:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"eb4cbcd3153b21738ac467dadf51583b106d94cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:38:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/51/659551/3/check/requirements-check/ca3cd72/ : SUCCESS in 4m 55s\n- tempest-full http://logs.openstack.org/51/659551/3/check/tempest-full/adba627/ : SUCCESS in 1h 58m 33s\n- tempest-full-py3 http://logs.openstack.org/51/659551/3/check/tempest-full-py3/3bc005d/ : SUCCESS in 1h 45m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/3/check/openstack-tox-lower-constraints/253b364/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/3/check/openstack-tox-pep8/ea1816a/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/3/check/openstack-tox-py27/b9b6f14/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/3/check/openstack-tox-py36/d5f8c46/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/3/check/openstack-tox-py37/0e00efb/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/51/659551/3/check/openstack-tox-docs/ec89e6c/html/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/3/check/build-openstack-releasenotes/c6ff937/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"604dfbefc557001ba096af72c3c841764ddf212e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 09:38:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"38295822edc93be99fca49b39905a399673a0e07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:28:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/51/659551/3/gate/requirements-check/eef2e45/ : SUCCESS in 3m 47s\n- tempest-full http://logs.openstack.org/51/659551/3/gate/tempest-full/6441ba1/ : SUCCESS in 1h 46m 50s\n- tempest-full-py3 http://logs.openstack.org/51/659551/3/gate/tempest-full-py3/f012b59/ : SUCCESS in 1h 39m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/659551/3/gate/openstack-tox-lower-constraints/72664d3/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/51/659551/3/gate/openstack-tox-pep8/00fb192/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py27/4204b34/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py36/1feca1c/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/51/659551/3/gate/openstack-tox-py37/bb4d21c/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/51/659551/3/gate/openstack-tox-docs/29dbeb6/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/51/659551/3/gate/build-openstack-releasenotes/5f48967/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"90385dd4a5ad951760cfe1a7babc9076f34be429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74cc8cfe13bd55513524e26018669d75a7c3872b","revisions":{"fff86e44f6d15a6098bc7650b39fb0c9924b076a":{"kind":"REWORK","_number":1,"created":"2019-05-16 12:41:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/659551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/51/659551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/51/659551/1"}}},"commit":{"parents":[{"commit":"0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441","subject":"Merge \"update git.openstack.org to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 12:40:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 12:41:14.000000000","tz":120},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start\nmarking it there as well.\n\nChange-Id: Ieaf9c77d62551352833a644ab22dff261c08bb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fff86e44f6d15a6098bc7650b39fb0c9924b076a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/fff86e44f6d15a6098bc7650b39fb0c9924b076a"}]},"branch":"refs/heads/master"},"5a84f55f8874ba51259eccca14867d35ee043250":{"kind":"NO_CHANGE","_number":2,"created":"2019-05-16 12:43:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/659551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/51/659551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/51/659551/2"}}},"commit":{"parents":[{"commit":"0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441","subject":"Merge \"update git.openstack.org to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 12:40:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 12:43:09.000000000","tz":120},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start\nmarking it there as well.\n\nChange-Id: Ieaf9c77d62551352833a644ab22dff261c08bb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/5a84f55f8874ba51259eccca14867d35ee043250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/5a84f55f8874ba51259eccca14867d35ee043250"}]},"branch":"refs/heads/master"},"74cc8cfe13bd55513524e26018669d75a7c3872b":{"kind":"REWORK","_number":3,"created":"2019-05-16 14:30:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/659551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/51/659551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/51/659551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/51/659551/3"}}},"commit":{"parents":[{"commit":"0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441","subject":"Merge \"update git.openstack.org to opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/0fa7d4a0656d0f9bea4cb92aa89ebdfa7cacc441"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 12:40:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-16 14:30:18.000000000","tz":120},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start\nmarking it there as well.\n\nChange-Id: Ieaf9c77d62551352833a644ab22dff261c08bb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/74cc8cfe13bd55513524e26018669d75a7c3872b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/74cc8cfe13bd55513524e26018669d75a7c3872b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
