)]}'
{"id":"openstack%2Fstevedore~667897","triplet_id":"openstack%2Fstevedore~master~I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","project":"openstack/stevedore","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-27 10:59:34.000000000","updated":"2019-06-27 15:26:32.000000000","submitted":"2019-06-27 15:26:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667897-1561649192058-f2ad0837","meta_rev_id":"ce5e37e2c6f69e7318a388cd0ba1b35dcd1391b5","_number":667897,"virtual_id_number":667897,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-06-27 15:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-27 12:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-27 13:03:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-27 13:03:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 12:59:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-27 13:03:53.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-06-27 15:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2478c7267edb26d3f397254d4402c5dfc4ceda8b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 10:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0b1bf1c516c3cad23e0ef0f4249d6bc815a8e9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/97/667897/1/check/tempest-full/0b13ad1/ : SUCCESS in 1h 46m 41s\n- tempest-full-py3 http://logs.openstack.org/97/667897/1/check/tempest-full-py3/fb056f8/ : SUCCESS in 1h 30m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667897/1/check/openstack-tox-lower-constraints/82a2175/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/97/667897/1/check/openstack-tox-pep8/3be8b19/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/97/667897/1/check/openstack-tox-py27/884b369/ : SUCCESS in 2m 34s\n- openstack-tox-py36 http://logs.openstack.org/97/667897/1/check/openstack-tox-py36/dd4f3a2/ : SUCCESS in 2m 36s\n- openstack-tox-py37 http://logs.openstack.org/97/667897/1/check/openstack-tox-py37/f5f9975/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/97/667897/1/check/openstack-tox-docs/33bb7a4/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes http://logs.openstack.org/97/667897/1/check/build-openstack-releasenotes/68ac115/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9b7c1c5841aab968fcabb2cf20c4fc9546bdffe4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 12:59:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"14d940e66fd44e05b00357084c053e8b47d2e592","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-27 13:03:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e516dc673a26d88821ab781815e0433258441637","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-27 13:04:04.000000000","message":"Patch Set 1:\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"7d43bf8ec3b40543df637d0c55fbd144ab093469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:04:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"721e7d8609ac692c29c455da10d049695c42791c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:26:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/97/667897/1/gate/tempest-full/02e6b28/ : SUCCESS in 1h 55m 31s\n- tempest-full-py3 http://logs.openstack.org/97/667897/1/gate/tempest-full-py3/f784af0/ : SUCCESS in 1h 49m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667897/1/gate/openstack-tox-lower-constraints/cbd68bd/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/97/667897/1/gate/openstack-tox-pep8/1a7f2ec/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/97/667897/1/gate/openstack-tox-py27/3c2028d/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/97/667897/1/gate/openstack-tox-py36/ba7ab03/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/97/667897/1/gate/openstack-tox-py37/c51e4ee/ : SUCCESS in 2m 45s\n- openstack-tox-docs http://logs.openstack.org/97/667897/1/gate/openstack-tox-docs/896b03b/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes http://logs.openstack.org/97/667897/1/gate/build-openstack-releasenotes/3d5692c/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c60d5bf20425f99152723c21c873f97b7738201e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:26:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"feac47186d79569bb564e18216e07f18ed5ccd8b","revisions":{"feac47186d79569bb564e18216e07f18ed5ccd8b":{"kind":"REWORK","_number":1,"created":"2019-06-27 10:59:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/97/667897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/97/667897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/667897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/667897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/97/667897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/97/667897/1"}}},"commit":{"parents":[{"commit":"74cc8cfe13bd55513524e26018669d75a7c3872b","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/74cc8cfe13bd55513524e26018669d75a7c3872b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:59:31.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/feac47186d79569bb564e18216e07f18ed5ccd8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/feac47186d79569bb564e18216e07f18ed5ccd8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
