)]}'
{"id":"openstack%2Fstevedore~667973","triplet_id":"openstack%2Fstevedore~stable%2Fstein~I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","project":"openstack/stevedore","branch":"stable/stein","topic":"bindep-fallback","hashtags":[],"change_id":"I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-27 16:09:03.000000000","updated":"2019-07-08 04:54:08.000000000","submitted":"2019-07-08 04:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667973-1562561648055-29425bd4","meta_rev_id":"6105909f40feeeb7f56d13e06cb1b530fcaec3a5","_number":667973,"virtual_id_number":667973,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 04:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:07:53.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-04 10:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 22:31:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:07:53.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 10:51:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-04 22:31:58.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-08 03:07:53.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 04:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71aad7b38f1efd15950bc09dec47e164c0d6ec25","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-27 16:09:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"190bcdfc62064063ce112eb032eb74790e7b3ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:57:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/667973/1/check/tempest-full/68c9a8b/ : SUCCESS in 1h 41m 53s\n- tempest-full-py3 http://logs.openstack.org/73/667973/1/check/tempest-full-py3/5b941d8/ : SUCCESS in 1h 31m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/667973/1/check/openstack-tox-lower-constraints/42842f7/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/73/667973/1/check/openstack-tox-pep8/73c2374/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/73/667973/1/check/openstack-tox-py27/eeb0f67/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/73/667973/1/check/openstack-tox-py35/8d08dba/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/73/667973/1/check/openstack-tox-py36/33d2d65/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/73/667973/1/check/openstack-tox-docs/4ff8b0b/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/73/667973/1/check/build-openstack-releasenotes/cea3719/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f9676f7c3a6d9e5542355cd700926cc2ff0becdf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 10:51:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f69fdd6b1ac461801fca21fe11305e4d986f4b50","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-07-04 22:31:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"965eced5dda50726753a86711da5a4f78219dc0a","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:07:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8a04633a3f305de953b5e2cae27057ab5eebf9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:08:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d04b6cc8879a270a0f5a221c48b88bb180538aee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:54:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/73/667973/1/gate/tempest-full/f096962/ : SUCCESS in 1h 40m 50s\n- tempest-full-py3 http://logs.openstack.org/73/667973/1/gate/tempest-full-py3/fc86440/ : SUCCESS in 1h 35m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/667973/1/gate/openstack-tox-lower-constraints/d232eee/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/73/667973/1/gate/openstack-tox-pep8/8972ff2/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/73/667973/1/gate/openstack-tox-py27/ad42cb8/ : SUCCESS in 2m 37s\n- openstack-tox-py35 http://logs.openstack.org/73/667973/1/gate/openstack-tox-py35/a23c9b4/ : SUCCESS in 2m 39s\n- openstack-tox-py36 http://logs.openstack.org/73/667973/1/gate/openstack-tox-py36/c13c180/ : SUCCESS in 2m 47s\n- openstack-tox-docs http://logs.openstack.org/73/667973/1/gate/openstack-tox-docs/6600c5b/html/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/73/667973/1/gate/build-openstack-releasenotes/d5f792c/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1b6067de900eed89e054c074fa7766e8e5a26b1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:54:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"caf65ed26066b5f7d13cc50e67f1df852414b2dc","revisions":{"caf65ed26066b5f7d13cc50e67f1df852414b2dc":{"kind":"REWORK","_number":1,"created":"2019-06-27 16:09:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/73/667973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/stevedore","ref":"refs/changes/73/667973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/73/667973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/73/667973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/stevedore refs/changes/73/667973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/stevedore refs/changes/73/667973/1"}}},"commit":{"parents":[{"commit":"9462fd2c71f2252a487e218f2bd106abd58c9b0e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/9462fd2c71f2252a487e218f2bd106abd58c9b0e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 10:57:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-27 16:08:32.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I5efddfc72e9dc0dd1b8b9e1e37708e1d30ef3c39\n(cherry picked from commit feac47186d79569bb564e18216e07f18ed5ccd8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/caf65ed26066b5f7d13cc50e67f1df852414b2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/stevedore/commit/caf65ed26066b5f7d13cc50e67f1df852414b2dc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
